Skip to content

feat: Convert emoji shortname to unicode on notification emails. - #31225

Merged
kodiakhq[bot] merged 8 commits into
developfrom
feat/emailEmoji
Jan 22, 2024
Merged

feat: Convert emoji shortname to unicode on notification emails.#31225
kodiakhq[bot] merged 8 commits into
developfrom
feat/emailEmoji

Conversation

@gabriellsh

@gabriellsh gabriellsh commented Dec 13, 2023

Copy link
Copy Markdown
Member

Proposed changes (including videos or screenshots)

Only unicode emoji are supported
Before:
image

After
image

Issue(s)

PROD-48

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Dec 13, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7695d52

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Dec 13, 2023

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Comparison is base (3fdf8ac) 59.89% compared to head (7695d52) 59.85%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31225      +/-   ##
===========================================
- Coverage    59.89%   59.85%   -0.04%     
===========================================
  Files         1792     1795       +3     
  Lines        34898    34933      +35     
  Branches      7265     7272       +7     
===========================================
+ Hits         20901    20910       +9     
- Misses       12400    12416      +16     
- Partials      1597     1607      +10     
Flag Coverage Δ
e2e 53.18% <ø> (-0.04%) ⬇️
unit 76.50% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

…/emailEmoji

* 'develop' of github.com:RocketChat/Rocket.Chat:
  chore: `Field` normalization (#31182)
  refactor: `Conversations` Overview Analytics (#31217)
  fix: Notifications for thread messages are sent even after thread has been read (#31181)
  ci: remove carryforward from codecov (#31216)
  fix: Wrong sysmessage when resuming chat from on hold (#31205)
  fix: Gallery not working inside threads (#31213)
@scuciatto scuciatto added this to the 6.6 milestone Jan 9, 2024
…/emailEmoji

* 'develop' of github.com:RocketChat/Rocket.Chat: (166 commits)
  feat: Add `chat.getURLPreview` endpoint (#30478)
  chore: QoL changes for Omnichannel (#31468)
  fix: add missing labels to UsersByTimeOfTheDaySection (#31368)
  refactor: Remove `Notifications` module (#31463)
  i18n: zh translation error (#21600)
  feat: `Bubble` on new messages indicators (#31478)
  test: fix flaky setGuestName test (#31486)
  Release 6.5.3
  chore: RateLimiter to async (#31480)
  chore: Revert "chore: Add missing changeset file" (#31481)
  chore: Remove `MessageContentBody` additional styles (#31476)
  chore: Replace `Notifications` module in favor of `SDK` inside CachedCollection (#31424)
  chore: improve backend startup (#31479)
  chore: fix embedded livechat tests (#31477)
  fix: Department name being double escaped (#31454)
  chore: Remove unnecessary changeset file (#31472)
  chore(eslint-config): Include missing ESLint configuration for React (#30977)
  refactor: Replace `Notifications` module in favor of `sdk` for Omnichannel/WebRTC (#31460)
  refactor: Replace `Notifications` module in favor of `sdk` for Video Conf (#31462)
  feat: Add support to cancel editing message action (#31347)
  ...
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Jan 22, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 22, 2024
@kodiakhq
kodiakhq Bot merged commit 9217c4f into develop Jan 22, 2024
@kodiakhq
kodiakhq Bot deleted the feat/emailEmoji branch January 22, 2024 14:51
gabriellsh added a commit that referenced this pull request Jan 23, 2024
* 'develop' of github.com:RocketChat/Rocket.Chat: (143 commits)
  fix: Marketplace apps installed as private showing as installed (#31514)
  test: fix messaging flaky test (#31512)
  fix: change the push sound sent when the push is from video conference (#30910)
  fix: ensure sessionId on Sessions documents (#31487)
  chore: handle rocket.cat creation and deletion (#31170)
  chore(agenda): Changes in Agenda API (#31427)
  refactor: Pay the debt of some TODO comments (#30338)
  feat: Room header keyboard navigability (#31516)
  fix: Cron job for clearing OEmbed cache isn't working (#31336)
  refactor(client): Handling of custom statuses (#31500)
  fix: Prevent `Fallback forward departments` feature to go on loop when configured as circular chain (#31328)
  feat: Add `push.test` endpoint (#31289)
  chore: split UserProvider into two: User and Authentication (#31513)
  fix: Can't remove channel's password to join (#31507)
  chore: convert CAS integration code to typescript (#31492)
  feat: Composer keyboard navigability (#31510)
  fix: Render warning in Logs - (MessageType.render is deprecated. Use MessageType.message instead) (#31426)
  feat: Convert emoji shortname to unicode on notification emails. (#31225)
  chore: Normalize `ButtonGroup` (#31499)
  chore: improve type definitions for login service configurations (#31491)
  ...
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants