fix: ensure sessionId on Sessions documents - #31487
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: f66e82c The changes in this PR will be included in the next version bump. This PR includes changesets to release 30 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov ReportAttention:
Additional details and impacted files@@ Coverage Diff @@
## develop #31487 +/- ##
===========================================
- Coverage 49.55% 49.55% -0.01%
===========================================
Files 3313 3310 -3
Lines 81476 81462 -14
Branches 16710 16711 +1
===========================================
- Hits 40378 40371 -7
+ Misses 36392 36385 -7
Partials 4706 4706
Flags with carried forward coverage won't be shown. Click here to find out more. |
sampaiodiego
requested changes
Jan 18, 2024
sampaiodiego
left a comment
Member
There was a problem hiding this comment.
what specific query is this index for? I didn't find one in the model class.. I wonder if this is for an "old" query we don't use anymore.
sampaiodiego
approved these changes
Jan 19, 2024
gabriellsh
added a commit
that referenced
this pull request
Jan 23, 2024
* 'develop' of github.com:RocketChat/Rocket.Chat: (143 commits) fix: Marketplace apps installed as private showing as installed (#31514) test: fix messaging flaky test (#31512) fix: change the push sound sent when the push is from video conference (#30910) fix: ensure sessionId on Sessions documents (#31487) chore: handle rocket.cat creation and deletion (#31170) chore(agenda): Changes in Agenda API (#31427) refactor: Pay the debt of some TODO comments (#30338) feat: Room header keyboard navigability (#31516) fix: Cron job for clearing OEmbed cache isn't working (#31336) refactor(client): Handling of custom statuses (#31500) fix: Prevent `Fallback forward departments` feature to go on loop when configured as circular chain (#31328) feat: Add `push.test` endpoint (#31289) chore: split UserProvider into two: User and Authentication (#31513) fix: Can't remove channel's password to join (#31507) chore: convert CAS integration code to typescript (#31492) feat: Composer keyboard navigability (#31510) fix: Render warning in Logs - (MessageType.render is deprecated. Use MessageType.message instead) (#31426) feat: Convert emoji shortname to unicode on notification emails. (#31225) chore: Normalize `ButtonGroup` (#31499) chore: improve type definitions for login service configurations (#31491) ...
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
Jira
Issue(s)
Steps to test or reproduce
Further comments