Skip to content

fix: Marketplace apps installed as private showing as installed - #31514

Merged
kodiakhq[bot] merged 3 commits into
developfrom
fix/private-apps-installed
Jan 23, 2024
Merged

fix: Marketplace apps installed as private showing as installed#31514
kodiakhq[bot] merged 3 commits into
developfrom
fix/private-apps-installed

Conversation

@tiagoevanp

@tiagoevanp tiagoevanp commented Jan 22, 2024

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

AECO-310

@tiagoevanp
tiagoevanp requested a review from a team as a code owner January 22, 2024 17:44
@changeset-bot

changeset-bot Bot commented Jan 22, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 50ee644

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Jan 22, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Comparison is base (1d81df0) 49.58% compared to head (50ee644) 49.56%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31514      +/-   ##
===========================================
- Coverage    49.58%   49.56%   -0.02%     
===========================================
  Files         3320     3320              
  Lines        81552    81552              
  Branches     16755    16755              
===========================================
- Hits         40436    40420      -16     
- Misses       36403    36422      +19     
+ Partials      4713     4710       -3     
Flag Coverage Δ
e2e 53.19% <ø> (-0.05%) ⬇️
e2e-api 40.68% <ø> (-0.02%) ⬇️
unit 76.47% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

csuadev
csuadev previously approved these changes Jan 22, 2024
@tiagoevanp tiagoevanp added this to the 6.6 milestone Jan 23, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Jan 23, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 23, 2024
@kodiakhq
kodiakhq Bot merged commit da410ef into develop Jan 23, 2024
@kodiakhq
kodiakhq Bot deleted the fix/private-apps-installed branch January 23, 2024 12:44
gabriellsh added a commit that referenced this pull request Jan 23, 2024
* 'develop' of github.com:RocketChat/Rocket.Chat: (143 commits)
  fix: Marketplace apps installed as private showing as installed (#31514)
  test: fix messaging flaky test (#31512)
  fix: change the push sound sent when the push is from video conference (#30910)
  fix: ensure sessionId on Sessions documents (#31487)
  chore: handle rocket.cat creation and deletion (#31170)
  chore(agenda): Changes in Agenda API (#31427)
  refactor: Pay the debt of some TODO comments (#30338)
  feat: Room header keyboard navigability (#31516)
  fix: Cron job for clearing OEmbed cache isn't working (#31336)
  refactor(client): Handling of custom statuses (#31500)
  fix: Prevent `Fallback forward departments` feature to go on loop when configured as circular chain (#31328)
  feat: Add `push.test` endpoint (#31289)
  chore: split UserProvider into two: User and Authentication (#31513)
  fix: Can't remove channel's password to join (#31507)
  chore: convert CAS integration code to typescript (#31492)
  feat: Composer keyboard navigability (#31510)
  fix: Render warning in Logs - (MessageType.render is deprecated. Use MessageType.message instead) (#31426)
  feat: Convert emoji shortname to unicode on notification emails. (#31225)
  chore: Normalize `ButtonGroup` (#31499)
  chore: improve type definitions for login service configurations (#31491)
  ...
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants