Skip to content

fix: Prevent Fallback forward departments feature to go on loop when configured as circular chain - #31328

Merged
kodiakhq[bot] merged 10 commits into
developfrom
fix/fallback-dep-loop
Jan 22, 2024
Merged

fix: Prevent Fallback forward departments feature to go on loop when configured as circular chain#31328
kodiakhq[bot] merged 10 commits into
developfrom
fix/fallback-dep-loop

Conversation

@KevLehman

@KevLehman KevLehman commented Dec 27, 2023

Copy link
Copy Markdown
Member

SUP-449

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Dec 27, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0cbfc7c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/model-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/rest-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/models Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/ddp-client Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Dec 27, 2023

Copy link
Copy Markdown

Codecov Report

Attention: 4 lines in your changes are missing coverage. Please review.

Comparison is base (7c6198f) 59.88% compared to head (0cbfc7c) 49.55%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff              @@
##           develop   #31328       +/-   ##
============================================
- Coverage    59.88%   49.55%   -10.34%     
============================================
  Files         1793     3315     +1522     
  Lines        34897    81475    +46578     
  Branches      7263    16736     +9473     
============================================
+ Hits         20898    40374    +19476     
- Misses       12398    36397    +23999     
- Partials      1601     4704     +3103     
Flag Coverage Δ
e2e 53.18% <ø> (-0.04%) ⬇️
e2e-api 40.67% <42.85%> (?)
unit 76.50% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Jan 11, 2024
@KevLehman
KevLehman marked this pull request as ready for review January 12, 2024 14:03
@KevLehman
KevLehman requested review from a team as code owners January 12, 2024 14:03
@KevLehman KevLehman added this to the 6.6 milestone Jan 16, 2024
lmauromb
lmauromb previously approved these changes Jan 16, 2024
@AliNunes AliNunes modified the milestones: 6.6, 6.5 Jan 16, 2024
@KevLehman KevLehman modified the milestones: 6.5, 6.6 Jan 19, 2024

@MarcosSpessatto MarcosSpessatto left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Would it be possible to add some test cases for the maximum number of depth?

Comment thread apps/meteor/ee/app/livechat-enterprise/server/hooks/onTransferFailure.ts Outdated
Comment thread apps/meteor/lib/callbacks.ts Outdated
Comment thread apps/meteor/ee/app/livechat-enterprise/server/hooks/onTransferFailure.ts Outdated
Comment thread apps/meteor/lib/callbacks.ts Outdated
@dionisio-bot dionisio-bot Bot removed the stat: ready to merge PR tested and approved waiting for merge label Jan 19, 2024
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Jan 19, 2024
@KevLehman

Copy link
Copy Markdown
Member Author

Tests are definitely possible. The test would take some time however to create the number of departments required for properly testing depth, so at the end I didn't do it 🤔

I can include it on a separate PR too.

@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 22, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 22, 2024
@kodiakhq
kodiakhq Bot merged commit b4b2cd2 into develop Jan 22, 2024
@kodiakhq
kodiakhq Bot deleted the fix/fallback-dep-loop branch January 22, 2024 21:23
gabriellsh added a commit that referenced this pull request Jan 23, 2024
* 'develop' of github.com:RocketChat/Rocket.Chat: (143 commits)
  fix: Marketplace apps installed as private showing as installed (#31514)
  test: fix messaging flaky test (#31512)
  fix: change the push sound sent when the push is from video conference (#30910)
  fix: ensure sessionId on Sessions documents (#31487)
  chore: handle rocket.cat creation and deletion (#31170)
  chore(agenda): Changes in Agenda API (#31427)
  refactor: Pay the debt of some TODO comments (#30338)
  feat: Room header keyboard navigability (#31516)
  fix: Cron job for clearing OEmbed cache isn't working (#31336)
  refactor(client): Handling of custom statuses (#31500)
  fix: Prevent `Fallback forward departments` feature to go on loop when configured as circular chain (#31328)
  feat: Add `push.test` endpoint (#31289)
  chore: split UserProvider into two: User and Authentication (#31513)
  fix: Can't remove channel's password to join (#31507)
  chore: convert CAS integration code to typescript (#31492)
  feat: Composer keyboard navigability (#31510)
  fix: Render warning in Logs - (MessageType.render is deprecated. Use MessageType.message instead) (#31426)
  feat: Convert emoji shortname to unicode on notification emails. (#31225)
  chore: Normalize `ButtonGroup` (#31499)
  chore: improve type definitions for login service configurations (#31491)
  ...
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants