Skip to content

feat: Bot warning when mentioning users not in a channel - #30464

Merged
kodiakhq[bot] merged 37 commits into
developfrom
feat/mentionBot
Dec 14, 2023
Merged

feat: Bot warning when mentioning users not in a channel#30464
kodiakhq[bot] merged 37 commits into
developfrom
feat/mentionBot

Conversation

@gabriellsh

@gabriellsh gabriellsh commented Sep 21, 2023

Copy link
Copy Markdown
Member

Proposed changes (including videos or screenshots)

Issue(s)

TC-948

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Sep 21, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 486954a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Minor
@rocket.chat/core-typings Minor
@rocket.chat/rest-typings Minor
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Major
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Major
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Major
@rocket.chat/models Patch
@rocket.chat/ui-client Major
@rocket.chat/ui-video-conf Major
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Major
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Sep 21, 2023

Copy link
Copy Markdown

Codecov Report

Merging #30464 (486954a) into develop (c407aaf) will decrease coverage by 28.13%.
The diff coverage is 21.12%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff              @@
##           develop   #30464       +/-   ##
============================================
- Coverage    77.40%   49.27%   -28.13%     
============================================
  Files          277     3297     +3020     
  Lines         8731    80908    +72177     
  Branches      1620    16508    +14888     
============================================
+ Hits          6758    39869    +33111     
- Misses        1565    36380    +34815     
- Partials       408     4659     +4251     
Flag Coverage Δ
e2e 52.58% <ø> (∅)
e2e-api 40.63% <21.12%> (?)
unit 77.40% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

…/mentionBot

* 'develop' of github.com:RocketChat/Rocket.Chat:
  chore: move Omnichannel toolbox section from sidebar room list (#30502)
  fix: RTL lang crashes Moderation Console  (#30393)
  feat: New records page analytics tab (#30373)
  chore: Changing some key translations - Setup Wizard (#30462)
  feat: use isolated-vm to run integration scripts (#30229)
  fix: do not broadcast events from the local node to the local service (duplicated event) (#30446)
  fix: Microsoft autotranslate not working (#30390)
  chore: `ResetPasswordPage` a11y improvements (#30479)
  chore: `ResetPasswordForm` a11y improvements (#30476)
  chore: Move bad words filter callback to service (#30241)
  fix: Message disappears from room after deletion even if "Show Deleted Status" is enabled (#30452)
  chore: add tooltip to mentions (#30445)
@scuciatto scuciatto added this to the 6.6.0 milestone Oct 4, 2023
gabriellsh and others added 10 commits October 9, 2023 13:01
…/mentionBot

* 'develop' of github.com:RocketChat/Rocket.Chat: (68 commits)
  fix: don't save the filter value on offlineMessageChannelName (#30576)
  feat: New permission to kick users (#30535)
  chore: improve message link focus (#30578)
  chore: Add required visual indication on `EditRoom` name input (#30589)
  fix: user dropdown menu position on RTL layout (#30490)
  chore: add missing `_id` field to `AtLeast` (#30592)
  test: Improve tests for legacy omnichannel analytics  (#30379)
  fix: Remove monitors query restrictions on update (#30550)
  chore: license v3 invalidation (#30585)
  chore: `CreateDirectMessage` a11y form improvements (#30581)
  chore: `CreateDiscussion` a11y improvements (#30583)
  chore: License v3 - behavior and limited reached triggers (#30561)
  chore: Improve `groups.create` endpoint for large amounts of members (#30499)
  feat: Monitors able to forward chats without joining (#30549)
  feat: Deployment Fingerprint (#30411)
  chore: fix flaky agents & custom fields tables tests (#30580)
  feat: Save visitor's activity on agent's interaction (#30222)
  chore: update mentions upsell modal content (#30503)
  fix: font-disabled color (#30569)
  chore: anchor text-decoration rules (#30485)
  ...
…/mentionBot

* 'develop' of github.com:RocketChat/Rocket.Chat:
  fix: Auto-translate doesn't work on E2E rooms (#30369)
  test: Validate monitors data access (#30593)
  fix: reset file input after uploading file (#30529)
  refactor: AutoCompleteTagsMultiple to TS (#30577)
  fix: DM room stay "read only" after the deactivate and reactivate user with "guest" role (#30459)
  fix: Change plan name from Enterprise to Premium on marketplace filtering (#30616)
  regression: test license from cloud (#30609)
  fix: Unable to send attachments via email as an omni-agent (#30525)
  ci: fix lint task
  fix: cloud alerts (#30607)
  feat: License trigger cloud sync after reach limits (#30603)
  chore: Change plan name Enterprise to Premium on marketplace (#30487)
  chore: Improve cache of static files (#30290)
  regression: Undefined MAC count on startup cloud sync (#30605)
  chore: Change Records page name to Reports (#30513)
…/mentionBot

* 'develop' of github.com:RocketChat/Rocket.Chat:
  feat: add tooltip to badge mentions (#30590)
  chore: improve `Tag` a11y link (#30636)
  refactor: Replace `useForm` in favor of RHF on `AppInstallPage` (#30634)
  fix: Improve FileProxy Handling, set Content-Type (#30427)
  refactor: `EditRoomInfo` to typescript (#28318)
  fix: mobile ringing notification missing call id (#30614)
  fix: Some HTTP requests sent by apps don't have their data parsed into JSON (#30560)
  test: More tests for groups kick (#30536)
  fix: Threads breaking after sending messages too fast (#30622)
  chore: Remove text decoration from room tag (#30606)
  i18n: Language update from LingoHub 🤖 on 2023-10-10Z (#30613)
  fix: File attachments have no content when exporting room messages as file (#30596)
  fix: use setImmediate to handle username update (#30500)
  chore: `AnalyticsReports` visual adequacy (#30617)
@gabriellsh
gabriellsh marked this pull request as ready for review October 19, 2023 14:31
@gabriellsh
gabriellsh requested review from a team as code owners October 19, 2023 14:31
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: conflict labels Nov 22, 2023
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Dec 1, 2023
@dionisio-bot dionisio-bot Bot removed the stat: ready to merge PR tested and approved waiting for merge label Dec 4, 2023
@scuciatto scuciatto added the stat: ready to merge PR tested and approved waiting for merge label Dec 14, 2023
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Dec 14, 2023
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Dec 14, 2023
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Dec 14, 2023
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Dec 14, 2023
@kodiakhq
kodiakhq Bot merged commit 44dd24d into develop Dec 14, 2023
@kodiakhq
kodiakhq Bot deleted the feat/mentionBot branch December 14, 2023 21:33
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants