chore: Improve cache of static files - #30290
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 2b5503d The changes in this PR will be included in the next version bump. This PR includes changesets to release 30 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov Report
@@ Coverage Diff @@
## develop #30290 +/- ##
========================================
Coverage 51.31% 51.32%
========================================
Files 811 805 -6
Lines 15081 15109 +28
Branches 2753 2763 +10
========================================
+ Hits 7739 7754 +15
+ Misses 6933 6930 -3
- Partials 409 425 +16
Flags with carried forward coverage won't be shown. Click here to find out more. |
gabriellsh
marked this pull request as ready for review
September 13, 2023 14:11
gabriellsh
marked this pull request as draft
September 13, 2023 18:09
gabriellsh
force-pushed
the
improve/ui-caching
branch
from
October 2, 2023 18:59
03d2e94 to
ca56388
Compare
sampaiodiego
requested changes
Oct 5, 2023
Co-authored-by: Diego Sampaio <chinello@gmail.com>
…t into improve/ui-caching * 'improve/ui-caching' of github.com:RocketChat/Rocket.Chat: (40 commits) Update .changeset/wicked-humans-hang.md feat: Save visitor's activity on agent's interaction (#30222) chore: update mentions upsell modal content (#30503) fix: font-disabled color (#30569) chore: anchor text-decoration rules (#30485) chore: `AccessibilityPage` hide Enterprise tag when is EE (#30486) chore: Replace `Field.[SUBCOMPONENT]` in favor of named imports (#30501) fix: Check for room scoped permissions for creating discussions (#30497) ci: fix Docker image build for production (#30562) chore: Fix triggers flaky tests (#30530) chore: get translations from CDN (#30331) feat: new `licenses.info` endpoint (#30473) ci: run tests from forks (#30556) regression: unmarked dangling promise on license validation (#30557) regression: fix initializing startup order (#30555) feat: push notification statistics (#30269) chore: do not focus messagebox on mobile devices (#30553) chore: add the new endpoints to sync with cloud (#30546) fix: in forward search field, user cannot be found by name (Full Name) (#29663) chore: set license public key v3 with v2 (#30548) ...
…ove/ui-caching * 'develop' of github.com:RocketChat/Rocket.Chat: fix: Remove monitors query restrictions on update (#30550) chore: license v3 invalidation (#30585) chore: `CreateDirectMessage` a11y form improvements (#30581) chore: `CreateDiscussion` a11y improvements (#30583) chore: License v3 - behavior and limited reached triggers (#30561) chore: Improve `groups.create` endpoint for large amounts of members (#30499) feat: Monitors able to forward chats without joining (#30549) feat: Deployment Fingerprint (#30411) chore: fix flaky agents & custom fields tables tests (#30580)
sampaiodiego
requested changes
Oct 6, 2023
sampaiodiego
approved these changes
Oct 9, 2023
sampaiodiego
pushed a commit
that referenced
this pull request
Oct 9, 2023
gabriellsh
added a commit
that referenced
this pull request
Oct 11, 2023
…/mentionBot * 'develop' of github.com:RocketChat/Rocket.Chat: fix: Auto-translate doesn't work on E2E rooms (#30369) test: Validate monitors data access (#30593) fix: reset file input after uploading file (#30529) refactor: AutoCompleteTagsMultiple to TS (#30577) fix: DM room stay "read only" after the deactivate and reactivate user with "guest" role (#30459) fix: Change plan name from Enterprise to Premium on marketplace filtering (#30616) regression: test license from cloud (#30609) fix: Unable to send attachments via email as an omni-agent (#30525) ci: fix lint task fix: cloud alerts (#30607) feat: License trigger cloud sync after reach limits (#30603) chore: Change plan name Enterprise to Premium on marketplace (#30487) chore: Improve cache of static files (#30290) regression: Undefined MAC count on startup cloud sync (#30605) chore: Change Records page name to Reports (#30513)
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
Issue(s)
Steps to test or reproduce
Further comments