Skip to content

fix: mobile ringing notification missing call id - #30614

Merged
kodiakhq[bot] merged 2 commits into
developfrom
fix/videoconf-ringing-missing-callid
Oct 13, 2023
Merged

fix: mobile ringing notification missing call id#30614
kodiakhq[bot] merged 2 commits into
developfrom
fix/videoconf-ringing-missing-callid

Conversation

@pierre-lehnen-rc

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Oct 10, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ec3e9d5

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pierre-lehnen-rc pierre-lehnen-rc added this to the 6.5.0 milestone Oct 10, 2023
@codecov

codecov Bot commented Oct 10, 2023

Copy link
Copy Markdown

Codecov Report

Merging #30614 (ec3e9d5) into develop (3536342) will decrease coverage by 3.86%.
The diff coverage is n/a.

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #30614      +/-   ##
===========================================
- Coverage    51.28%   47.42%   -3.86%     
===========================================
  Files          811      796      -15     
  Lines        15074    14881     -193     
  Branches      2752     2766      +14     
===========================================
- Hits          7731     7058     -673     
- Misses        6935     7399     +464     
- Partials       408      424      +16     
Flag Coverage Δ
e2e 44.10% <ø> (-4.46%) ⬇️
unit 64.80% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@scuciatto
scuciatto marked this pull request as ready for review October 11, 2023 17:00
@scuciatto
scuciatto requested a review from a team as a code owner October 11, 2023 17:00
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Oct 13, 2023
@scuciatto scuciatto modified the milestones: 6.5.0, 6.4.2 Oct 13, 2023
@kodiakhq
kodiakhq Bot merged commit dd254a9 into develop Oct 13, 2023
@kodiakhq
kodiakhq Bot deleted the fix/videoconf-ringing-missing-callid branch October 13, 2023 14:34
gabriellsh added a commit that referenced this pull request Oct 16, 2023
…/mentionBot

* 'develop' of github.com:RocketChat/Rocket.Chat:
  feat: add tooltip to badge mentions (#30590)
  chore: improve `Tag` a11y link (#30636)
  refactor: Replace `useForm` in favor of RHF on `AppInstallPage` (#30634)
  fix: Improve FileProxy Handling, set Content-Type (#30427)
  refactor: `EditRoomInfo` to typescript (#28318)
  fix: mobile ringing notification missing call id (#30614)
  fix: Some HTTP requests sent by apps don't have their data parsed into JSON (#30560)
  test: More tests for groups kick (#30536)
  fix: Threads breaking after sending messages too fast (#30622)
  chore: Remove text decoration from room tag (#30606)
  i18n: Language update from LingoHub 🤖 on 2023-10-10Z (#30613)
  fix: File attachments have no content when exporting room messages as file (#30596)
  fix: use setImmediate to handle username update (#30500)
  chore: `AnalyticsReports` visual adequacy (#30617)
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants