fix: DM room stay "read only" after the deactivate and reactivate user with "guest" role - #30459
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 6a3b2cb The changes in this PR will be included in the next version bump. This PR includes changesets to release 30 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov Report
@@ Coverage Diff @@
## develop #30459 +/- ##
===========================================
- Coverage 51.30% 51.05% -0.25%
===========================================
Files 811 806 -5
Lines 15083 15116 +33
Branches 2754 2765 +11
===========================================
- Hits 7738 7718 -20
- Misses 6935 6970 +35
- Partials 410 428 +18
Flags with carried forward coverage won't be shown. Click here to find out more. |
heitortanoue
marked this pull request as ready for review
September 26, 2023 16:41
matheusbsilva137
approved these changes
Sep 29, 2023
tassoevan
approved these changes
Oct 10, 2023
gabriellsh
added a commit
that referenced
this pull request
Oct 11, 2023
…/mentionBot * 'develop' of github.com:RocketChat/Rocket.Chat: fix: Auto-translate doesn't work on E2E rooms (#30369) test: Validate monitors data access (#30593) fix: reset file input after uploading file (#30529) refactor: AutoCompleteTagsMultiple to TS (#30577) fix: DM room stay "read only" after the deactivate and reactivate user with "guest" role (#30459) fix: Change plan name from Enterprise to Premium on marketplace filtering (#30616) regression: test license from cloud (#30609) fix: Unable to send attachments via email as an omni-agent (#30525) ci: fix lint task fix: cloud alerts (#30607) feat: License trigger cloud sync after reach limits (#30603) chore: Change plan name Enterprise to Premium on marketplace (#30487) chore: Improve cache of static files (#30290) regression: Undefined MAC count on startup cloud sync (#30605) chore: Change Records page name to Reports (#30513)
hugocostadev
pushed a commit
that referenced
this pull request
Oct 11, 2023
…r with "guest" role (#30459) Co-authored-by: Matheus Barbosa Silva <36537004+matheusbsilva137@users.noreply.github.com>
debdutdeb
pushed a commit
that referenced
this pull request
Oct 26, 2023
…r with "guest" role (#30459) Co-authored-by: Matheus Barbosa Silva <36537004+matheusbsilva137@users.noreply.github.com>
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
This bug is with the direct message between user and guest.
If we deactivate and reactivate user, his permission in direct message chat with guest users will be in a read-only mode.
Issue(s)
Steps to test or reproduce
Create a normal user and a guest user. Initiate chat between them so there is a direct room.
Now deactivate and activate the normal user.
Room will become and stay read-only even after reactivation of guest user.
This is not happening in DM with User<->User
Further comments
SUP-221