Skip to content

fix: Change plan name from Enterprise to Premium on marketplace filtering - #30616

Merged
kodiakhq[bot] merged 2 commits into
developfrom
fix/marketplace-premium-category
Oct 10, 2023
Merged

fix: Change plan name from Enterprise to Premium on marketplace filtering#30616
kodiakhq[bot] merged 2 commits into
developfrom
fix/marketplace-premium-category

Conversation

@csuadev

@csuadev csuadev commented Oct 10, 2023

Copy link
Copy Markdown
Contributor

Due to the marketplace don't showing apps when Enterprise category is selected, change category name on filtered apps since a change on the database from all apps from Enterprise to Premium.

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Oct 10, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9ca9356

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@csuadev
csuadev marked this pull request as ready for review October 10, 2023 15:59
@csuadev
csuadev requested a review from a team as a code owner October 10, 2023 15:59
@codecov

codecov Bot commented Oct 10, 2023

Copy link
Copy Markdown

Codecov Report

Merging #30616 (9ca9356) into develop (b7ba752) will decrease coverage by 0.02%.
The diff coverage is n/a.

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #30616      +/-   ##
===========================================
- Coverage    51.30%   51.29%   -0.02%     
===========================================
  Files          811      805       -6     
  Lines        15083    15112      +29     
  Branches      2754     2765      +11     
===========================================
+ Hits          7738     7751      +13     
  Misses        6935     6935              
- Partials       410      426      +16     
Flag Coverage Δ
e2e 48.62% <ø> (+0.04%) ⬆️
unit 64.80% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@yash-rajpal yash-rajpal changed the title fix: Change plan name from Enterprise to Premium on marketplace filtering chore: Change plan name from Enterprise to Premium on marketplace filtering Oct 10, 2023
@yash-rajpal yash-rajpal changed the title chore: Change plan name from Enterprise to Premium on marketplace filtering fix: Change plan name from Enterprise to Premium on marketplace filtering Oct 10, 2023
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Oct 10, 2023
@casalsgh casalsgh modified the milestones: 6.4.2, 6.5.0 Oct 10, 2023
@kodiakhq
kodiakhq Bot merged commit 0586501 into develop Oct 10, 2023
@kodiakhq
kodiakhq Bot deleted the fix/marketplace-premium-category branch October 10, 2023 16:53
@casalsgh

Copy link
Copy Markdown
Contributor

AECO-409 is the one related to this

gabriellsh added a commit that referenced this pull request Oct 11, 2023
…/mentionBot

* 'develop' of github.com:RocketChat/Rocket.Chat:
  fix: Auto-translate doesn't work on E2E rooms (#30369)
  test: Validate monitors data access (#30593)
  fix: reset file input after uploading file (#30529)
  refactor: AutoCompleteTagsMultiple to TS (#30577)
  fix: DM room stay "read only" after the deactivate and reactivate user with "guest" role (#30459)
  fix: Change plan name from Enterprise to Premium on marketplace filtering (#30616)
  regression: test license from cloud (#30609)
  fix: Unable to send attachments via email as an omni-agent (#30525)
  ci: fix lint task
  fix: cloud alerts (#30607)
  feat: License trigger cloud sync after reach limits (#30603)
  chore: Change plan name Enterprise to Premium on marketplace (#30487)
  chore: Improve cache of static files (#30290)
  regression: Undefined MAC count on startup cloud sync (#30605)
  chore: Change Records page name to Reports (#30513)
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants