OCPBUGS-20097: Migrate tools image to RHEL9 - #1652
Conversation
|
@soltysh: This pull request references Jira Issue OCPBUGS-20097, which is invalid:
Comment The bug has been updated to refer to the pull request using the external bug tracker. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
/lgtm |
|
/jira refresh |
|
@soltysh: This pull request references Jira Issue OCPBUGS-20097, which is valid. The bug has been moved to the POST state. 3 validation(s) were run on this bug
DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: ardaguclu, soltysh The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/hold |
|
|
we need to fix our builder, just looking where we're missing |
Good to know that this is something we can fix our own. Thanks for the answer. |
|
/test e2e-aws-ovn |
|
/test images |
|
@soltysh: This pull request references Jira Issue OCPBUGS-20097. The bug has been updated to no longer refer to the pull request using the external bug tracker. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
unrelated; |
|
reminder: Bot reverted openshift/release#50563, we need to reintroduce our changes before retesting. |
|
Opened this openshift/release#50581 |
To go with openshift/oc#1652
|
/retest |
|
/retest-required |
|
There needs to be a rhel9 change in oc in release repo which was reverted by bot. |
To go with openshift/oc#1652
|
/test all |
|
/hold |
|
/lgtm |
|
This proofed to work |
|
/override ci/prow/e2e-aws-ovn-serial |
|
@soltysh: Overrode contexts on behalf of soltysh: ci/prow/e2e-aws-ovn-serial DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
@soltysh: The following test failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
|
@soltysh: Jira Issue OCPBUGS-20097: All pull requests linked via external trackers have merged:
Jira Issue OCPBUGS-20097 has been moved to the MODIFIED state. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
To go with openshift/oc#1652
|
[ART PR BUILD NOTIFIER] This PR has been included in build openshift-enterprise-cli-container-v4.16.0-202404221711.p0.g057cfc5.assembly.stream.el8 for distgit openshift-enterprise-cli. |
/assign @ardaguclu