Skip to content

OCPBUGS-20097: Update oc master configuration to RHEL9 - #50468

Merged
openshift-merge-bot[bot] merged 3 commits into
openshift:masterfrom
ardaguclu:define-rhel9-for-oc
Apr 3, 2024
Merged

OCPBUGS-20097: Update oc master configuration to RHEL9#50468
openshift-merge-bot[bot] merged 3 commits into
openshift:masterfrom
ardaguclu:define-rhel9-for-oc

Conversation

@ardaguclu

Copy link
Copy Markdown
Member

Opened this #49421 to support both RHEL9 and RHEL8 base images. However, bot reverted it da07fce#diff-a3aabdc6d9228089f6578c4e764fc60a076217ec3a31ca778e4df1f652432de1 by removing the unused tags.

This PR reintroduces this but directly changing the base images to RHEL9, so that bot will not remove them again.

Problem in previous PR was that tests image was using RHEL8 base image which seems to be coming from https://github.com/openshift/origin/blob/master/images/tests/Dockerfile.rhel. I'm not sure how we can fix this chicken-egg problem. Because we are doing this to bump tests image to RHEL9.

@openshift-ci-robot openshift-ci-robot added jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. labels Apr 2, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@ardaguclu: This pull request references Jira Issue OCPBUGS-20097, which is valid.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.16.0) matches configured target version for branch (4.16.0)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

Opened this #49421 to support both RHEL9 and RHEL8 base images. However, bot reverted it da07fce#diff-a3aabdc6d9228089f6578c4e764fc60a076217ec3a31ca778e4df1f652432de1 by removing the unused tags.

This PR reintroduces this but directly changing the base images to RHEL9, so that bot will not remove them again.

Problem in previous PR was that tests image was using RHEL8 base image which seems to be coming from https://github.com/openshift/origin/blob/master/images/tests/Dockerfile.rhel. I'm not sure how we can fix this chicken-egg problem. Because we are doing this to bump tests image to RHEL9.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested review from mfojtik and soltysh April 2, 2024 11:45
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 2, 2024
@ardaguclu

Copy link
Copy Markdown
Member Author

/pj-rehearse

@ardaguclu

Copy link
Copy Markdown
Member Author

/pj-rehearse

@ardaguclu

Copy link
Copy Markdown
Member Author

/test pull-ci-openshift-oc-release-4.16-images

@openshift-ci

openshift-ci Bot commented Apr 2, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: The specified target(s) for /test were not found.
The following commands are available to trigger required jobs:

  • /test app-ci-config-dry
  • /test boskos-config
  • /test boskos-config-generation
  • /test build-clusters
  • /test build01-dry
  • /test build02-dry
  • /test build03-dry
  • /test build04-dry
  • /test build05-dry
  • /test build09-dry
  • /test build10-dry
  • /test check-gh-automation
  • /test check-gh-automation-tide
  • /test ci-operator-config
  • /test ci-operator-config-metadata
  • /test ci-operator-registry
  • /test ci-secret-bootstrap-config-validation
  • /test ci-testgrid-allow-list
  • /test clusterimageset-validate
  • /test config
  • /test core-valid
  • /test deprecate-templates
  • /test generated-config
  • /test generated-dashboards
  • /test hive-dry
  • /test jira-lifecycle-config
  • /test openshift-image-mirror-mappings
  • /test ordered-prow-config
  • /test owners
  • /test pr-reminder-config
  • /test prow-config
  • /test prow-config-filenames
  • /test prow-config-semantics
  • /test pylint
  • /test release-config
  • /test release-controller-config
  • /test secret-generator-config-valid
  • /test services-valid
  • /test stackrox-stackrox-stackrox-stackrox-check
  • /test step-registry-metadata
  • /test step-registry-shellcheck
  • /test supplemental-ci-images-mapping-files
  • /test sync-rover-groups
  • /test vsphere02-dry
  • /test yamllint

Use /test all to run the following jobs that were automatically triggered:

  • pull-ci-openshift-release-check-gh-automation
  • pull-ci-openshift-release-check-gh-automation-tide
  • pull-ci-openshift-release-master-build-clusters
  • pull-ci-openshift-release-master-ci-operator-config
  • pull-ci-openshift-release-master-ci-operator-config-metadata
  • pull-ci-openshift-release-master-ci-operator-registry
  • pull-ci-openshift-release-master-core-valid
  • pull-ci-openshift-release-master-deprecate-templates
  • pull-ci-openshift-release-master-generated-config
  • pull-ci-openshift-release-master-ordered-prow-config
  • pull-ci-openshift-release-master-owners
  • pull-ci-openshift-release-master-prow-config-filenames
  • pull-ci-openshift-release-master-release-controller-config
  • pull-ci-openshift-release-openshift-image-mirror-mappings
  • pull-ci-openshift-release-yamllint
Details

In response to this:

/test pull-ci-openshift-oc-release-4.16-images

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@ardaguclu

Copy link
Copy Markdown
Member Author

/pj-rehearse pull-ci-openshift-oc-master-e2e-agnostic-ovn-cmd
/pj-rehearse pull-ci-openshift-oc-release-4.16-images

@ardaguclu

Copy link
Copy Markdown
Member Author

/pj-rehearse pull-ci-openshift-oc-release-4.16-images

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@ardaguclu: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-oc-release-4.16-build-rpms-from-tar openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.16-e2e-agnostic-ovn-cmd openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.16-e2e-aws-certrotation openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.16-e2e-aws-ovn openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.16-e2e-aws-ovn-builds openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.16-e2e-aws-ovn-serial openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.16-e2e-aws-ovn-upgrade openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.16-e2e-metal-ipi-ovn-ipv6 openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.16-images openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.16-rpm-build openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.16-security openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.16-unit openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.16-verify openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.16-verify-deps openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-build-rpms-from-tar openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-agnostic-ovn-cmd openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-aws-certrotation openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-aws-ovn openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-aws-ovn-builds openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-aws-ovn-serial openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-aws-ovn-upgrade openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-metal-ipi-ovn-ipv6 openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-images openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-rpm-build openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-security openshift/oc presubmit Ci-operator config changed

A total of 42 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse abort to abort all active rehearsals

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@soltysh soltysh left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@soltysh

soltysh commented Apr 3, 2024

Copy link
Copy Markdown
Contributor

/pj-rehearse ack

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Apr 3, 2024
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 3, 2024
@openshift-ci

openshift-ci Bot commented Apr 3, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ardaguclu, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Apr 3, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/rehearse/openshift/oc/release-4.17/e2e-aws-ovn 1779369 link unknown /pj-rehearse pull-ci-openshift-oc-release-4.17-e2e-aws-ovn
ci/rehearse/openshift/oc/release-4.17/e2e-aws-ovn-builds 1779369 link unknown /pj-rehearse pull-ci-openshift-oc-release-4.17-e2e-aws-ovn-builds
ci/rehearse/openshift/oc/release-4.17/e2e-aws-certrotation 1779369 link unknown /pj-rehearse pull-ci-openshift-oc-release-4.17-e2e-aws-certrotation

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 0c7c07b into openshift:master Apr 3, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@ardaguclu: Jira Issue OCPBUGS-20097: Some pull requests linked via external trackers have merged:

The following pull requests linked via external trackers have not merged:

These pull request must merge or be unlinked from the Jira bug in order for it to move to the next state. Once unlinked, request a bug refresh with /jira refresh.

Jira Issue OCPBUGS-20097 has not been moved to the MODIFIED state.

Details

In response to this:

Opened this #49421 to support both RHEL9 and RHEL8 base images. However, bot reverted it da07fce#diff-a3aabdc6d9228089f6578c4e764fc60a076217ec3a31ca778e4df1f652432de1 by removing the unused tags.

This PR reintroduces this but directly changing the base images to RHEL9, so that bot will not remove them again.

Problem in previous PR was that tests image was using RHEL8 base image which seems to be coming from https://github.com/openshift/origin/blob/master/images/tests/Dockerfile.rhel. I'm not sure how we can fix this chicken-egg problem. Because we are doing this to bump tests image to RHEL9.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@ardaguclu
ardaguclu deleted the define-rhel9-for-oc branch April 3, 2024 09:23
BradLugo pushed a commit to BradLugo/openshift-release that referenced this pull request Jun 13, 2024
* Update oc master configuration to RHEL9

* Update 4.16 and 4.16 oc configurations

* Remove redundant base image and use base-rhel9
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants