[e2e] bump generated key size - #2153
Conversation
This change increases the size of the generated private keys in the e2e test suite so that it complies with the smaller size of keys in recent openssl version. See https://access.redhat.com/documentation/en-us/red_hat_enterprise_linux/9/html-single/9.3_release_notes/index
|
Skipping CI for Draft Pull Request. |
|
/test azure-e2e-operator |
|
/approve |
|
I logged in the CI cluster and the e2e passed the issue |
|
/test remaining-required |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: jrvaldes, sebsoto The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/lgtm |
|
/retest-required |
|
@jrvaldes: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
This change increases the size of the generated private keys in the e2e test suite so that it complies with the smaller size of keys in recent openssl version.
See https://access.redhat.com/documentation/en-us/red_hat_enterprise_linux/9/html-single/9.3_release_notes/index