OCPBUGS-20097: Use RHEL8 oc explicitly - #4006
Conversation
|
@ardaguclu: This pull request explicitly references no jira issue. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
/test remaining-required |
|
seems unrelated failure; |
|
multi_stage tests are passing but e2e failed in pod_scaler test. I don't think, it is related to this PR but we need to ensure. |
|
As seen in this job https://prow.ci.openshift.org/view/gs/test-platform-results/pr-logs/pull/openshift_ci-tools/4007/pull-ci-openshift-ci-tools-master-e2e/1764611836759511040 that was triggered by dummy PR #4007, e2e job is broken. |
|
/retest-required |
|
e2e test successfully passed against 4.13 images https://prow.ci.openshift.org/view/gs/test-platform-results/pr-logs/pull/openshift_ci-tools/4006/pull-ci-openshift-ci-tools-master-e2e/1764620755875663872 which proves that fall back mechanism works against older versions. |
|
/test remaining-required |
|
e2e test successfully passed against 4.15 image https://prow.ci.openshift.org/view/gs/test-platform-results/pr-logs/pull/openshift_ci-tools/4006/pull-ci-openshift-ci-tools-master-e2e/1764653723851689984 which proves that the patch works against clusters >= 4.15 |
|
@ardaguclu: This pull request explicitly references no jira issue. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
@ardaguclu: This pull request references Jira Issue OCPBUGS-20097, which is valid. 3 validation(s) were run on this bug
The bug has been updated to refer to the pull request using the external bug tracker. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
/lgtm |
|
@danilo-gemoli thanks for tagging. Before unholding, I'd like to have review from @soltysh to lower the possibility of issues that happened in #4000 |
soltysh
left a comment
There was a problem hiding this comment.
/lgtm
probably worth re-trying the merges after the shift week
|
/lgtm |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: ardaguclu, danilo-gemoli, joepvd, soltysh The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/hold cancel |
|
@ardaguclu: The following test failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
|
@ardaguclu: Jira Issue OCPBUGS-20097: Some pull requests linked via external trackers have merged: The following pull requests linked via external trackers have not merged:
These pull request must merge or be unlinked from the Jira bug in order for it to move to the next state. Once unlinked, request a bug refresh with Jira Issue OCPBUGS-20097 has not been moved to the MODIFIED state. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
Previous PR #4000 was reverted in #4002 because changes were not backwards compatible and caused cascading failures.
This PR reintroduces same approach additionally supporting backwards compatibility. Firstly, it checks the path of
oc.rhel8. If binary is found, it will copy. If binary is not found like releases older than 4.15, it falls back to current behavior.Thanks to this change, maintainers of this repository can change rhel version of oc and also supports multiple architecture versions (i.e. linux_arm64).
Edited: I'd like to note the reason of this change is that we are going to change the base image of oc from RHEL8 to RHEL9 and oc compiled in RHEL9 does not work on RHEL8 clusters. Therefore, we have to explicitly use oc RHEL8 binary in ci-tools, until ci-tools is decided to bumped to RHEL9.