OCPBUGS-20097: Migrate tools image to RHEL9 - #28733
Conversation
|
@soltysh: This pull request references Jira Issue OCPBUGS-20097, which is valid. 3 validation(s) were run on this bug
The bug has been updated to refer to the pull request using the external bug tracker. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
/lgtm |
| RUN PACKAGES="git gzip util-linux" && \ | ||
| if [ $HOSTTYPE = x86_64 ]; then PACKAGES="$PACKAGES python3-cinderclient"; fi && \ | ||
| if [ $HOSTTYPE = x86_64 ]; then PACKAGES="$PACKAGES rt-tests rteval"; fi && \ | ||
| if [ $HOSTTYPE = x86_64 ]; then PACKAGES="$PACKAGES realtime-tests rteval"; fi && \ |
There was a problem hiding this comment.
The CI reports No match for argument: realtime-tests. So this rpm is not available?
There was a problem hiding this comment.
Wrong base image is being used, it's rhel8, and should be rhel9
|
With openshift/release#51253 in |
|
/test all |
|
Job Failure Risk Analysis for sha: 9aa4be9
|
|
/test all |
|
/lgtm |
|
/test all |
|
/test all |
|
Job Failure Risk Analysis for sha: 87ef7d2
|
The package rt-tests is now called realtime-tests
|
/lgtm |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: ardaguclu, soltysh The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
I think, we have a sufficient signal to merge this. |
|
@soltysh: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
|
/hold cancel |
|
@soltysh: Jira Issue OCPBUGS-20097: All pull requests linked via external trackers have merged:
Jira Issue OCPBUGS-20097 has been moved to the MODIFIED state. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
[ART PR BUILD NOTIFIER] This PR has been included in build openshift-enterprise-tests-container-v4.16.0-202404241818.p0.gce9d443.assembly.stream.el9 for distgit openshift-enterprise-tests. |
/assign @ardaguclu