Skip to content

NO-JIRA: Reintroduce tools rhel9 base images - #51176

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
ardaguclu:readd-rhel9-oc
Apr 22, 2024
Merged

NO-JIRA: Reintroduce tools rhel9 base images#51176
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
ardaguclu:readd-rhel9-oc

Conversation

@ardaguclu

Copy link
Copy Markdown
Member

SSIA

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Apr 22, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@ardaguclu: This pull request explicitly references no jira issue.

Details

In response to this:

SSIA

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested review from deads2k and mfojtik April 22, 2024 04:59
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 22, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@ardaguclu: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-oc-master-build-rpms-from-tar openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-agnostic-ovn-cmd openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-aws-certrotation openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-aws-ovn openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-aws-ovn-builds openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-aws-ovn-serial openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-aws-ovn-upgrade openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-metal-ipi-ovn-ipv6 openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-images openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-rpm-build openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-security openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-unit openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-verify openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-verify-deps openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-build-rpms-from-tar openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-agnostic-ovn-cmd openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-aws-certrotation openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-aws-ovn openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-aws-ovn-builds openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-aws-ovn-serial openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-aws-ovn-upgrade openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-e2e-metal-ipi-ovn-ipv6 openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-images openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-rpm-build openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-release-4.17-security openshift/oc presubmit Ci-operator config changed

A total of 42 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here
Prior to this PR being merged, you will need to either run and acknowledge or opt to skip these rehearsals.

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse abort to abort all active rehearsals

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@soltysh soltysh left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@soltysh

soltysh commented Apr 22, 2024

Copy link
Copy Markdown
Contributor

/pj-rehearse skip

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@soltysh: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Apr 22, 2024
@soltysh

soltysh commented Apr 22, 2024

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 22, 2024
@openshift-ci

openshift-ci Bot commented Apr 22, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ardaguclu, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Apr 22, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 0c27a86 into openshift:master Apr 22, 2024
Patryk-Stefanski pushed a commit to Patryk-Stefanski/release that referenced this pull request Apr 24, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants