Skip to content

OCPBUGS-20097: Switch to use cli image, as the scratch image - #28695

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
soltysh:switch_images
Apr 19, 2024
Merged

OCPBUGS-20097: Switch to use cli image, as the scratch image#28695
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
soltysh:switch_images

Conversation

@soltysh

@soltysh soltysh commented Apr 8, 2024

Copy link
Copy Markdown
Contributor

/assign @ardaguclu

Additional followup to #28687

@openshift-ci-robot openshift-ci-robot added jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. labels Apr 8, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@soltysh: This pull request references Jira Issue OCPBUGS-20097, which is valid.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.16.0) matches configured target version for branch (4.16.0)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

/assign @ardaguclu

Additional followup to #28687

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@soltysh

soltysh commented Apr 8, 2024

Copy link
Copy Markdown
Contributor Author

This is blocking openshift/oc#1652

@ardaguclu

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 8, 2024
@openshift-ci

openshift-ci Bot commented Apr 8, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ardaguclu, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 8, 2024
@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 1972e66 and 2 for PR HEAD 3dc2100 in total

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 7a9d6c2 and 1 for PR HEAD 3dc2100 in total

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 856924d and 0 for PR HEAD 3dc2100 in total

@openshift-ci-robot

Copy link
Copy Markdown

/hold

Revision 3dc2100 was retested 3 times: holding

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 9, 2024
@soltysh

soltysh commented Apr 11, 2024

Copy link
Copy Markdown
Contributor Author

/retest-required

@soltysh

soltysh commented Apr 11, 2024

Copy link
Copy Markdown
Contributor Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 11, 2024
@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD f38832a and 2 for PR HEAD 3dc2100 in total

@soltysh

soltysh commented Apr 11, 2024

Copy link
Copy Markdown
Contributor Author

/test e2e-metal-ipi-ovn-ipv6

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 15cebbf and 1 for PR HEAD 3dc2100 in total

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD abe5694 and 0 for PR HEAD 3dc2100 in total

@openshift-ci-robot

Copy link
Copy Markdown

/hold

Revision 3dc2100 was retested 3 times: holding

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 12, 2024
@joepvd

joepvd commented Apr 15, 2024

Copy link
Copy Markdown
Contributor

/retest

@soltysh

soltysh commented Apr 16, 2024

Copy link
Copy Markdown
Contributor Author

/retest-required

@soltysh

soltysh commented Apr 17, 2024

Copy link
Copy Markdown
Contributor Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 17, 2024
@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD f1a21f4 and 2 for PR HEAD 3dc2100 in total

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD f2d3513 and 1 for PR HEAD 3dc2100 in total

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 3d40ef3 and 0 for PR HEAD 3dc2100 in total

@openshift-ci-robot

Copy link
Copy Markdown

/hold

Revision 3dc2100 was retested 3 times: holding

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 17, 2024
@soltysh

soltysh commented Apr 18, 2024

Copy link
Copy Markdown
Contributor Author

/test e2e-gcp-ovn-builds

@soltysh

soltysh commented Apr 18, 2024

Copy link
Copy Markdown
Contributor Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 18, 2024
@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 886b7ae and 2 for PR HEAD 3dc2100 in total

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD e5c2630 and 1 for PR HEAD 3dc2100 in total

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 5ac6743 and 0 for PR HEAD 3dc2100 in total

@openshift-ci

openshift-ci Bot commented Apr 18, 2024

Copy link
Copy Markdown
Contributor

@soltysh: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-ovn-single-node-serial 3dc2100 link false /test e2e-aws-ovn-single-node-serial
ci/prow/e2e-aws-ovn-single-node-upgrade 3dc2100 link false /test e2e-aws-ovn-single-node-upgrade

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-ci-robot

Copy link
Copy Markdown

/hold

Revision 3dc2100 was retested 3 times: holding

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 18, 2024
@soltysh

soltysh commented Apr 19, 2024

Copy link
Copy Markdown
Contributor Author

/hold cancel

@soltysh

soltysh commented Apr 19, 2024

Copy link
Copy Markdown
Contributor Author

/override ci/prow/e2e-gcp-ovn-builds
this is being fixed in #28718

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 19, 2024
@openshift-ci

openshift-ci Bot commented Apr 19, 2024

Copy link
Copy Markdown
Contributor

@soltysh: Overrode contexts on behalf of soltysh: ci/prow/e2e-gcp-ovn-builds

Details

In response to this:

/override ci/prow/e2e-gcp-ovn-builds
this is being fixed in #28718

/hold cancel

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD eb64483 and 2 for PR HEAD 3dc2100 in total

@openshift-trt-bot

Copy link
Copy Markdown

Job Failure Risk Analysis for sha: 3dc2100

Job Name Failure Risk
pull-ci-openshift-origin-master-e2e-aws-ovn-single-node-serial Low
[sig-arch] events should not repeat pathologically for ns/openshift-etcd-operator
This test has passed 50.79% of 63 runs on jobs ['periodic-ci-openshift-release-master-nightly-4.16-e2e-aws-ovn-single-node-serial'] in the last 14 days.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 34cefb1 into openshift:master Apr 19, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@soltysh: Jira Issue OCPBUGS-20097: Some pull requests linked via external trackers have merged:

The following pull requests linked via external trackers have not merged:

These pull request must merge or be unlinked from the Jira bug in order for it to move to the next state. Once unlinked, request a bug refresh with /jira refresh.

Jira Issue OCPBUGS-20097 has not been moved to the MODIFIED state.

Details

In response to this:

/assign @ardaguclu

Additional followup to #28687

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

This PR has been included in build openshift-enterprise-tests-container-v4.16.0-202404191609.p0.g34cefb1.assembly.stream.el8 for distgit openshift-enterprise-tests.
All builds following this will include this PR.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants