Skip to content

WRKLDS-1251: Switch cli image to RHEL9 base image - #1752

Merged
openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
ardaguclu:cli-switch-rhel9
May 4, 2024
Merged

WRKLDS-1251: Switch cli image to RHEL9 base image#1752
openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
ardaguclu:cli-switch-rhel9

Conversation

@ardaguclu

@ardaguclu ardaguclu commented Apr 29, 2024

Copy link
Copy Markdown
Member

This PR bumps the base image of cli image to RHEL9. This is the last remaining piece after the bump of tools image in #1652.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Apr 29, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@ardaguclu: This pull request explicitly references no jira issue.

Details

In response to this:

SSIA

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@ardaguclu
ardaguclu marked this pull request as draft April 29, 2024 06:03
@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Apr 29, 2024
@openshift-ci
openshift-ci Bot requested review from deads2k and mfojtik April 29, 2024 06:03
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 29, 2024
@ardaguclu
ardaguclu marked this pull request as ready for review April 29, 2024 06:22
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Apr 29, 2024
@openshift-ci
openshift-ci Bot requested a review from soltysh April 29, 2024 06:22
@ardaguclu ardaguclu closed this Apr 29, 2024
@ardaguclu ardaguclu reopened this Apr 29, 2024
@ardaguclu

Copy link
Copy Markdown
Member Author

/test all

@ardaguclu ardaguclu changed the title NO-JIRA: Switch cli image to RHEL9 base image WRKLDS-1251: Switch cli image to RHEL9 base image Apr 30, 2024
@openshift-ci-robot

openshift-ci-robot commented Apr 30, 2024

Copy link
Copy Markdown

@ardaguclu: This pull request references WRKLDS-1251 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the task to target the "4.16.0" version, but no target version was set.

Details

In response to this:

This PR bumps the base image of cli image to RHEL9. This is the last remaining piece after the bump of tools image in #1652.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

4 similar comments
@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

5 similar comments
@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

joepvd added a commit to joepvd/ocp-build-data that referenced this pull request May 3, 2024
joepvd added a commit to joepvd/ocp-build-data that referenced this pull request May 3, 2024
joepvd added a commit to joepvd/ocp-build-data that referenced this pull request May 3, 2024
@joepvd

joepvd commented May 3, 2024

Copy link
Copy Markdown
Contributor

/payload 4.16 nightly blocking

@openshift-ci

openshift-ci Bot commented May 3, 2024

Copy link
Copy Markdown
Contributor

@joepvd: trigger 8 job(s) of type blocking for the nightly release of OCP 4.16

  • periodic-ci-openshift-release-master-nightly-4.16-e2e-aws-sdn-upgrade
  • periodic-ci-openshift-release-master-ci-4.16-e2e-azure-ovn-upgrade
  • periodic-ci-openshift-release-master-ci-4.16-upgrade-from-stable-4.15-e2e-gcp-ovn-rt-upgrade
  • periodic-ci-openshift-hypershift-release-4.16-periodics-e2e-aws-ovn-conformance
  • periodic-ci-openshift-release-master-nightly-4.16-e2e-aws-ovn-serial
  • periodic-ci-openshift-release-master-ci-4.16-e2e-aws-ovn-upgrade
  • periodic-ci-openshift-release-master-nightly-4.16-e2e-metal-ipi-ovn-bm
  • periodic-ci-openshift-release-master-nightly-4.16-e2e-metal-ipi-ovn-ipv6

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/644be180-0921-11ef-9246-13b045bcbecb-0

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

1 similar comment
@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

https://pr-payload-tests.ci.openshift.org/runs/ci/644be180-0921-11ef-9246-13b045bcbecb-0 Contains some failures looking like flakes;
/payload 4.16 nightly blocking

@openshift-ci

openshift-ci Bot commented May 3, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: trigger 8 job(s) of type blocking for the nightly release of OCP 4.16

  • periodic-ci-openshift-release-master-nightly-4.16-e2e-aws-sdn-upgrade
  • periodic-ci-openshift-release-master-ci-4.16-e2e-azure-ovn-upgrade
  • periodic-ci-openshift-release-master-ci-4.16-upgrade-from-stable-4.15-e2e-gcp-ovn-rt-upgrade
  • periodic-ci-openshift-hypershift-release-4.16-periodics-e2e-aws-ovn-conformance
  • periodic-ci-openshift-release-master-nightly-4.16-e2e-aws-ovn-serial
  • periodic-ci-openshift-release-master-ci-4.16-e2e-aws-ovn-upgrade
  • periodic-ci-openshift-release-master-nightly-4.16-e2e-metal-ipi-ovn-bm
  • periodic-ci-openshift-release-master-nightly-4.16-e2e-metal-ipi-ovn-ipv6

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/10a22570-0945-11ef-86d2-6eb13b7413db-0

@ardaguclu

Copy link
Copy Markdown
Member Author

/retest

@ardaguclu

Copy link
Copy Markdown
Member Author

I'm not sure why the test pods were killed unexpectedly;
/payload 4.16 nightly blocking

@openshift-ci

openshift-ci Bot commented May 3, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: trigger 8 job(s) of type blocking for the nightly release of OCP 4.16

  • periodic-ci-openshift-release-master-nightly-4.16-e2e-aws-sdn-upgrade
  • periodic-ci-openshift-release-master-ci-4.16-e2e-azure-ovn-upgrade
  • periodic-ci-openshift-release-master-ci-4.16-upgrade-from-stable-4.15-e2e-gcp-ovn-rt-upgrade
  • periodic-ci-openshift-hypershift-release-4.16-periodics-e2e-aws-ovn-conformance
  • periodic-ci-openshift-release-master-nightly-4.16-e2e-aws-ovn-serial
  • periodic-ci-openshift-release-master-ci-4.16-e2e-aws-ovn-upgrade
  • periodic-ci-openshift-release-master-nightly-4.16-e2e-metal-ipi-ovn-bm
  • periodic-ci-openshift-release-master-nightly-4.16-e2e-metal-ipi-ovn-ipv6

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/7c71f8e0-095c-11ef-8f30-504bf3777d32-0

@ardaguclu

Copy link
Copy Markdown
Member Author

/payload 4.16 nightly blocking

@openshift-ci

openshift-ci Bot commented May 3, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: trigger 8 job(s) of type blocking for the nightly release of OCP 4.16

  • periodic-ci-openshift-release-master-nightly-4.16-e2e-aws-sdn-upgrade
  • periodic-ci-openshift-release-master-ci-4.16-e2e-azure-ovn-upgrade
  • periodic-ci-openshift-release-master-ci-4.16-upgrade-from-stable-4.15-e2e-gcp-ovn-rt-upgrade
  • periodic-ci-openshift-hypershift-release-4.16-periodics-e2e-aws-ovn-conformance
  • periodic-ci-openshift-release-master-nightly-4.16-e2e-aws-ovn-serial
  • periodic-ci-openshift-release-master-ci-4.16-e2e-aws-ovn-upgrade
  • periodic-ci-openshift-release-master-nightly-4.16-e2e-metal-ipi-ovn-bm
  • periodic-ci-openshift-release-master-nightly-4.16-e2e-metal-ipi-ovn-ipv6

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/7988e710-097a-11ef-879a-61d72a607b0f-0

@ardaguclu

Copy link
Copy Markdown
Member Author

/payload periodic-ci-openshift-hypershift-release-4.16-periodics-e2e-aws-ovn-conformance

@openshift-ci

openshift-ci Bot commented May 4, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: it appears that you have attempted to use some version of the payload command, but your comment was incorrectly formatted and cannot be acted upon. See the docs for usage info.

@ardaguclu

Copy link
Copy Markdown
Member Author

/payload-job periodic-ci-openshift-hypershift-release-4.16-periodics-e2e-aws-ovn-conformance

@openshift-ci

openshift-ci Bot commented May 4, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: trigger 1 job(s) for the /payload-(with-prs|job|aggregate|job-with-prs|aggregate-with-prs) command

  • periodic-ci-openshift-hypershift-release-4.16-periodics-e2e-aws-ovn-conformance

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/7f675590-09d4-11ef-8736-b9b8a36f77af-0

@joepvd

joepvd commented May 4, 2024

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 4, 2024
@openshift-ci

openshift-ci Bot commented May 4, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ardaguclu, joepvd

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit 941fe49 into openshift:master May 4, 2024
openshift-cherrypick-robot pushed a commit to openshift-cherrypick-robot/ocp-build-data that referenced this pull request May 4, 2024
@ardaguclu
ardaguclu deleted the cli-switch-rhel9 branch May 4, 2024 09:27
@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

This PR has been included in build openshift-enterprise-cli-container-v4.16.0-202405041048.p0.g941fe49.assembly.stream.el9 for distgit openshift-enterprise-cli.
All builds following this will include this PR.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants