Skip to content

Tools to rhel9 - #4598

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift-eng:openshift-4.16from
joepvd:tools-to-rhel9
Apr 22, 2024
Merged

Tools to rhel9#4598
openshift-merge-bot[bot] merged 1 commit into
openshift-eng:openshift-4.16from
joepvd:tools-to-rhel9

Conversation

@joepvd

@joepvd joepvd commented Apr 3, 2024

Copy link
Copy Markdown
Contributor

No description provided.

@openshift-bot

Copy link
Copy Markdown

Validating 3 file(s)...
Validating images/openshift-enterprise-tests.yml...
Traceback (most recent call last):
File "/usr/local/bin/validate-ocp-build-data", line 33, in
sys.exit(load_entry_point('rh-ocp-build-data-validator==0.2.5', 'console_scripts', 'validate-ocp-build-data')())
File "/usr/local/lib/python3.9/site-packages/validator/main.py", line 87, in main
validate(f, args.exclude_vpn, args.schema_only)
File "/usr/local/lib/python3.9/site-packages/validator/main.py", line 48, in validate
support.fail_validation(msg, parsed)
File "/usr/local/lib/python3.9/site-packages/validator/support.py", line 14, in fail_validation
raise exceptions.ValidationFailed(msg)
validator.exceptions.ValidationFailed: CGit validation failed for images/openshift-enterprise-tests.yml (https://pkgs.devel.redhat.com/cgit/containers/openshift-enterprise-tests)
Returned error: Branch rhaos-4.16-rhel-9 not found in CGit cache

@joepvd

joepvd commented Apr 3, 2024

Copy link
Copy Markdown
Contributor Author

/hold

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 3, 2024
@openshift-bot

Copy link
Copy Markdown

Validating 4 file(s)...
✅ Validated group.yml
Validating images/openshift-enterprise-tests.yml...
✅ Validated images/openshift-enterprise-tests.yml
Validating images/ose-network-tools.yml...
Traceback (most recent call last):
File "/usr/local/bin/validate-ocp-build-data", line 33, in
sys.exit(load_entry_point('rh-ocp-build-data-validator==0.2.5', 'console_scripts', 'validate-ocp-build-data')())
File "/usr/local/lib/python3.9/site-packages/validator/main.py", line 87, in main
validate(f, args.exclude_vpn, args.schema_only)
File "/usr/local/lib/python3.9/site-packages/validator/main.py", line 48, in validate
support.fail_validation(msg, parsed)
File "/usr/local/lib/python3.9/site-packages/validator/support.py", line 14, in fail_validation
raise exceptions.ValidationFailed(msg)
validator.exceptions.ValidationFailed: CGit validation failed for images/ose-network-tools.yml (https://pkgs.devel.redhat.com/cgit/containers/ose-network-tools)
Returned error: Branch rhaos-4.16-rhel-9 not found in CGit cache

@openshift-merge-robot openshift-merge-robot added needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. and removed needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. labels Apr 13, 2024
@joepvd

joepvd commented Apr 22, 2024

Copy link
Copy Markdown
Contributor Author

/cherrypick openshift-4.17

@openshift-cherrypick-robot

Copy link
Copy Markdown

@joepvd: once the present PR merges, I will cherry-pick it on top of openshift-4.17 in a new PR and assign it to you.

Details

In response to this:

/cherrypick openshift-4.17

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@soltysh soltysh left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/hold cancel
openshift/oc#1652 just merged

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 22, 2024
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 22, 2024
@ashwindasr

Copy link
Copy Markdown
Contributor

/approve

@openshift-ci

openshift-ci Bot commented Apr 22, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ashwindasr, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 22, 2024
@openshift-merge-bot
openshift-merge-bot Bot merged commit 74be058 into openshift-eng:openshift-4.16 Apr 22, 2024
@openshift-cherrypick-robot

Copy link
Copy Markdown

@joepvd: new pull request created: #4705

Details

In response to this:

/cherrypick openshift-4.17

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants