Skip to content

OCPBUGS-20097: Switch tools image to rhel9 - #47663

Closed
soltysh wants to merge 1 commit into
openshift:masterfrom
soltysh:tools_rhel9
Closed

OCPBUGS-20097: Switch tools image to rhel9#47663
soltysh wants to merge 1 commit into
openshift:masterfrom
soltysh:tools_rhel9

Conversation

@soltysh

@soltysh soltysh commented Jan 15, 2024

Copy link
Copy Markdown
Contributor

Needed to land openshift/oc#1652

/assign @ardaguclu

@openshift-ci-robot openshift-ci-robot added jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. labels Jan 15, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@soltysh: This pull request references Jira Issue OCPBUGS-20097, which is valid.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.16.0) matches configured target version for branch (4.16.0)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

Needed to land openshift/oc#1652

/assign @ardaguclu

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@soltysh, pj-rehearse: unable to determine affected jobs. This could be due to a branch that needs to be rebased. ERROR:

could not load configuration from candidate revision of release repo: failed to load ci-operator configuration from release repo: failed to load ci-operator config (error converting YAML to JSON: yaml: line 48: mapping values are not allowed in this context)
Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse abort to abort all active rehearsals

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@soltysh

soltysh commented Jan 15, 2024

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ci/prow/e2e-agnostic-ovn-cmd

@openshift-ci

openshift-ci Bot commented Jan 15, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jan 15, 2024
@soltysh

soltysh commented Jan 15, 2024

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ci/prow/e2e-agnostic-ovn-cmd

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@soltysh, pj-rehearse: unable to determine affected jobs ERROR:

could not load configuration from candidate revision of release repo: failed to load ci-operator configuration from release repo: failed to load ci-operator config (error converting YAML to JSON: yaml: line 48: mapping values are not allowed in this context)

If the problem persists, please contact Test Platform.

@soltysh

soltysh commented Jan 15, 2024

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ci/prow/e2e-agnostic-ovn-cmd

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@soltysh: job(s): ci/prow/e2e-agnostic-ovn-cmd either don't exist or were not found to be affected, and cannot be rehearsed

@soltysh

soltysh commented Jan 15, 2024

Copy link
Copy Markdown
Contributor Author

/hold

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jan 15, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@soltysh: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-oc-master-build-rpms-from-tar openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-agnostic-ovn-cmd openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-aws-certrotation openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-aws-ovn openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-aws-ovn-builds openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-aws-ovn-serial openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-aws-ovn-upgrade openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-e2e-metal-ipi-ovn-ipv6 openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-images openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-rpm-build openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-security openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-unit openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-verify openshift/oc presubmit Ci-operator config changed
pull-ci-openshift-oc-master-verify-deps openshift/oc presubmit Ci-operator config changed
Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse abort to abort all active rehearsals

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@soltysh: job(s): ci/prow/e2e-agnostic-ovn-cmd either don't exist or were not found to be affected, and cannot be rehearsed

@openshift-bot

Copy link
Copy Markdown
Contributor

Issues in openshift/release go stale after 30d of inactivity.

Mark the issue as fresh by commenting /remove-lifecycle stale.
Stale issues rot after an additional 15d of inactivity and eventually close.
Exclude this issue from closing by commenting /lifecycle frozen.

If this issue is safe to close now please do so with /close.

/lifecycle stale

@openshift-ci openshift-ci Bot added the lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. label Feb 14, 2024
@openshift-bot

Copy link
Copy Markdown
Contributor

Stale issue in openshift/release rot after 15d of inactivity.

Mark the issue as fresh by commenting /remove-lifecycle rotten.
Rotten issues close after an additional 15d of inactivity.
Exclude this issue from closing by commenting /lifecycle frozen.

If this issue is safe to close now please do so with /close.

/lifecycle rotten
/remove-lifecycle stale

@openshift-ci openshift-ci Bot added lifecycle/rotten Denotes an issue or PR that has aged beyond stale and will be auto-closed. and removed lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. labels Feb 29, 2024
@ardaguclu

Copy link
Copy Markdown
Member

Opened this #49421 as alternative.

@openshift-ci

openshift-ci Bot commented Mar 1, 2024

Copy link
Copy Markdown
Contributor

@soltysh: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-bot

Copy link
Copy Markdown
Contributor

Rotten issues in openshift/release close after 15d of inactivity.

Reopen the issue by commenting /reopen.
Mark the issue as fresh by commenting /remove-lifecycle rotten.
Exclude this issue from closing again by commenting /lifecycle frozen.

/close

@openshift-ci openshift-ci Bot closed this Mar 16, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@soltysh: This pull request references Jira Issue OCPBUGS-20097. The bug has been updated to no longer refer to the pull request using the external bug tracker.

Details

In response to this:

Needed to land openshift/oc#1652

/assign @ardaguclu

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci

openshift-ci Bot commented Mar 16, 2024

Copy link
Copy Markdown
Contributor

@openshift-bot: Closed this PR.

Details

In response to this:

Rotten issues in openshift/release close after 15d of inactivity.

Reopen the issue by commenting /reopen.
Mark the issue as fresh by commenting /remove-lifecycle rotten.
Exclude this issue from closing again by commenting /lifecycle frozen.

/close

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@soltysh
soltysh deleted the tools_rhel9 branch April 24, 2024 13:08
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lifecycle/rotten Denotes an issue or PR that has aged beyond stale and will be auto-closed.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants