Skip to content

NO-ISSUE: Update Konflux references (cloud_hotfix_releases) - #6596

Closed
red-hat-konflux[bot] wants to merge 25 commits into
cloud_hotfix_releasesfrom
konflux/references/konflux-references
Closed

red-hat-konflux[bot] wants to merge 25 commits into
cloud_hotfix_releasesfrom
konflux/references/konflux-references

Conversation

@red-hat-konflux

@red-hat-konflux red-hat-konflux Bot commented Jul 20, 2024 •

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Notes
quay.io/konflux-ci/tekton-catalog/task-buildah 0.1 -> 0.2 ⚠️migration⚠️
quay.io/konflux-ci/tekton-catalog/task-source-build 5e34db0 -> 9eee3cf

/lgtm

/approve


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test

…ed (#6122)

Signed-off-by: Riccardo Piccoli <rpiccoli@redhat.com>
Co-authored-by: Riccardo Piccoli <rpiccoli@redhat.com>
…y_latest=true in .../openshift-versions (#6138)

Co-authored-by: danmanor <dmanor@redhat.com>
…run when the instance becomes leader instead of skipping if it not (#6136)

Co-authored-by: danmanor <dmanor@redhat.com>
Signed-off-by: Riccardo Piccoli <rpiccoli@redhat.com>
Co-authored-by: Riccardo Piccoli <rpiccoli@redhat.com>
…if MCE is not selected (#6160)

Signed-off-by: Riccardo Piccoli <rpiccoli@redhat.com>
Co-authored-by: Riccardo Piccoli <rpiccoli@redhat.com>
…Release Syncer - if there are release images already in the DB, continue using the stale data instead of failing / truncating the table (#6166)

Co-authored-by: danmanor <dmanor@redhat.com>
…itecture of both release and OS images to comply with ABI current behavior (#6204)

Co-authored-by: danmanor <dmanor@redhat.com>
…age using major.minor OpenShift version to latest non-beta release image if exists, or latest beta release image otherwise (#6205)

Co-authored-by: danmanor <dmanor@redhat.com>
…ons endpoint (#6206)

Co-authored-by: danmanor <dmanor@redhat.com>
…Release Syncer - if there are release images already in the DB, continue using the stale data instead of failing / truncating the table (#6145) (#6210)

Co-authored-by: Dan Manor <dmanor@redhat.com>
Open installation of OCP on iSCSI boot volume to all platforms.

Remove `ip=ibft` from the kernel arguments as it prevent proper network
configuration when several NICs are in use. RHCOS team recommend to
leave DHCP configuration.

Co-authored-by: Adrien Gentil <agentil@redhat.com>
…ng to pre-release suffix as well instead of just base version (#6226)

Co-authored-by: danmanor <dmanor@redhat.com>
… to retrieve multiarch release image in case a multi-arch OpenShift version, single-arch requested and it exist. (#6229)

Co-authored-by: danmanor <dmanor@redhat.com>
…PI flow to correctly compare pre-release versions (#6286)

Co-authored-by: danmanor <dmanor@redhat.com>
…orrect (#6303)

This PR addresses a few bugs in docs, regexes and code that cause inconsistency in the handling of patch manifests.

Patch manifests should end with the extension `.patch` or should end with a suffixed patch extension such as `.patch_extension`

Co-authored-by: Paul Maidment <pmaidmen@redhat.com>
Remove OCI from technology preview as it is now a supported integration.
The control-plane-machine-set (CPMS) cluster operator fails to deploy
on 4.16-ec.5 build for ABI deployments on vSphere.
That build removed the CPMS feature gate.

Because ABI and assisted deployments do not create or include
machines in the manifests, and the control plane machine operator
expects machines to be present for each control plane node, the
CO fails to deploy.

The CPMS manifest should be removed for ABI and other UPI-like
deployments.

For UPI there is documentation indicating the CPMS manifest must
be removed: https://docs.openshift.com/container-platform/4.15/installing/installing_vsphere/upi/installing-vsphere.html#installation-user-infra-generate-k8s-manifest-ignition_installing-vsphere

Co-authored-by: Richard Su <rwsu@redhat.com>
* Red Hat Konflux update assisted-service-saas-hotfix

* ci: Fix konflux git clone params (#6463)

---------

Co-authored-by: red-hat-konflux <konflux@no-reply.konflux-ci.dev>
Co-authored-by: Matthieu Bernardin <40832495+pastequo@users.noreply.github.com>
Signed-off-by: red-hat-konflux <123456+red-hat-konflux[bot]@users.noreply.github.com>
Co-authored-by: red-hat-konflux[bot] <126015336+red-hat-konflux[bot]@users.noreply.github.com>
Signed-off-by: red-hat-konflux <123456+red-hat-konflux[bot]@users.noreply.github.com>
Co-authored-by: red-hat-konflux[bot] <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@openshift-ci

openshift-ci Bot commented Jul 20, 2024

Copy link
Copy Markdown
Contributor

@red-hat-konflux[bot]: you cannot LGTM your own PR.

Details

In response to this:

This PR contains the following updates:

Package Change Notes
quay.io/konflux-ci/tekton-catalog/task-apply-tags 175162b -> 5168758
quay.io/konflux-ci/tekton-catalog/task-buildah 0.1 -> 0.2 ⚠️migration⚠️
quay.io/konflux-ci/tekton-catalog/task-clair-scan f38e274 -> a13278c
quay.io/konflux-ci/tekton-catalog/task-clamav-scan 46fcff2 -> a574202
quay.io/konflux-ci/tekton-catalog/task-deprecated-image-check d70d970 -> ea275ae
quay.io/konflux-ci/tekton-catalog/task-git-clone 9ea6101 -> de0ca88
quay.io/konflux-ci/tekton-catalog/task-init 83b7df5 -> 99c98d3
quay.io/konflux-ci/tekton-catalog/task-prefetch-dependencies c22f253 -> 9f1dd11
quay.io/konflux-ci/tekton-catalog/task-sast-snyk-check a74dae0 -> d68390c
quay.io/konflux-ci/tekton-catalog/task-sbom-json-check 5e0f1de -> acc9cb8
quay.io/konflux-ci/tekton-catalog/task-show-sbom 69edfd6 -> 7f8b549
quay.io/konflux-ci/tekton-catalog/task-source-build eacb3f4 -> 9eee3cf
quay.io/konflux-ci/tekton-catalog/task-summary 4b0563b -> d97c04a

/lgtm

/approve


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-ci-robot

Copy link
Copy Markdown

@red-hat-konflux[bot]: This pull request explicitly references no jira issue.

Details

In response to this:

This PR contains the following updates:

Package Change Notes
quay.io/konflux-ci/tekton-catalog/task-apply-tags 175162b -> 5168758
quay.io/konflux-ci/tekton-catalog/task-buildah 0.1 -> 0.2 ⚠️migration⚠️
quay.io/konflux-ci/tekton-catalog/task-clair-scan f38e274 -> a13278c
quay.io/konflux-ci/tekton-catalog/task-clamav-scan 46fcff2 -> a574202
quay.io/konflux-ci/tekton-catalog/task-deprecated-image-check d70d970 -> ea275ae
quay.io/konflux-ci/tekton-catalog/task-git-clone 9ea6101 -> de0ca88
quay.io/konflux-ci/tekton-catalog/task-init 83b7df5 -> 99c98d3
quay.io/konflux-ci/tekton-catalog/task-prefetch-dependencies c22f253 -> 9f1dd11
quay.io/konflux-ci/tekton-catalog/task-sast-snyk-check a74dae0 -> d68390c
quay.io/konflux-ci/tekton-catalog/task-sbom-json-check 5e0f1de -> acc9cb8
quay.io/konflux-ci/tekton-catalog/task-show-sbom 69edfd6 -> 7f8b549
quay.io/konflux-ci/tekton-catalog/task-source-build eacb3f4 -> 9eee3cf
quay.io/konflux-ci/tekton-catalog/task-summary 4b0563b -> d97c04a

/lgtm

/approve


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Jul 20, 2024
Signed-off-by: red-hat-konflux <123456+red-hat-konflux[bot]@users.noreply.github.com>
@red-hat-konflux red-hat-konflux Bot changed the title NO-ISSUE: Update Konflux references (master) NO-ISSUE: Update Konflux references (cloud_hotfix_releases) Jul 20, 2024
@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/references/konflux-references branch from 31f6fbc to 48f8215 Compare July 20, 2024 00:06
@red-hat-konflux
red-hat-konflux Bot changed the base branch from master to cloud_hotfix_releases July 20, 2024 00:06
@openshift-ci openshift-ci Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Jul 20, 2024
@openshift-ci

openshift-ci Bot commented Jul 20, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: red-hat-konflux[bot]
Once this PR has been reviewed and has the lgtm label, please assign javipolo for approval. For more information see the Kubernetes Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@codecov

codecov Bot commented Jul 20, 2024 •

Copy link
Copy Markdown

Codecov Report

Attention: Patch coverage is 58.25243% with 86 lines in your changes missing coverage. Please review.

Project coverage is 68.21%. Comparing base (fe16303) to head (48f8215).
Report is 148 commits behind head on cloud_hotfix_releases.

❗ Current head 48f8215 differs from pull request most recent head b5190b4

Please upload reports for the commit b5190b4 to get more accurate results.

Additional details and impacted files

Impacted file tree graph

@@                    Coverage Diff                    @@
##           cloud_hotfix_releases    #6596      +/-   ##
=========================================================
- Coverage                  68.32%   68.21%   -0.11%     
=========================================================
  Files                        242      242              
  Lines                      35786    35869      +83     
=========================================================
+ Hits                       24449    24469      +20     
- Misses                      9214     9276      +62     
- Partials                    2123     2124       +1     
Files Coverage Δ
internal/controller/controllers/images.go 100.00% <100.00%> (ø)
internal/featuresupport/features_platforms.go 92.65% <100.00%> (ø)
internal/hardware/validator.go 77.17% <100.00%> (+0.44%) ⬆️
internal/host/hostcommands/install_cmd.go 84.45% <100.00%> (ø)
internal/host/validator.go 81.73% <100.00%> (ø)
internal/manifests/manifests.go 78.49% <100.00%> (ø)
internal/operators/cnv/cnv_operator.go 79.65% <100.00%> (ø)
internal/operators/manager.go 78.07% <100.00%> (+1.65%) ⬆️
internal/operators/mce/mce_operator.go 71.55% <100.00%> (ø)
internal/operators/odf/validations.go 97.61% <100.00%> (ø)
... and 12 more

... and 1 file with indirect coverage changes

@red-hat-konflux

Copy link
Copy Markdown
Contributor Author

Edited/Blocked Notification

Renovate will not automatically rebase this PR, because it does not recognize the last commit author and assumes somebody else may have edited the PR.

You can manually request rebase by checking the rebase/retry box above.

⚠️ Warning: custom changes will be lost.

@adriengentil
adriengentil force-pushed the cloud_hotfix_releases branch from c8f8b96 to c5d796f Compare July 25, 2024 09:48
@openshift-merge-robot openshift-merge-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Jul 25, 2024
@openshift-merge-robot

Copy link
Copy Markdown

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-ci openshift-ci Bot added size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. api-review Categorizes an issue or PR as actively needing an API review. and removed size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Jul 25, 2024
@openshift-ci

openshift-ci Bot commented Jul 25, 2024

Copy link
Copy Markdown
Contributor

@red-hat-konflux[bot]: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/edge-images 48f8215 link true /test edge-images
ci/prow/edge-ci-index 48f8215 link true /test edge-ci-index
ci/prow/mce-images 48f8215 link true /test mce-images
ci/prow/e2e-metal-assisted-4-13 b5190b4 link false /test e2e-metal-assisted-4-13
ci/prow/e2e-metal-assisted-ipv4v6 b5190b4 link true /test e2e-metal-assisted-ipv4v6
ci/prow/e2e-metal-assisted-odf b5190b4 link true /test e2e-metal-assisted-odf
ci/prow/e2e-metal-assisted-single-node b5190b4 link true /test e2e-metal-assisted-single-node
ci/prow/subsystem-aws b5190b4 link true /test subsystem-aws
ci/prow/unit-test b5190b4 link true /test unit-test
ci/prow/images b5190b4 link true /test images
ci/prow/e2e-metal-assisted-cnv b5190b4 link true /test e2e-metal-assisted-cnv
ci/prow/e2e-metal-assisted-4-12 b5190b4 link true /test e2e-metal-assisted-4-12
ci/prow/verify-generated-code b5190b4 link true /test verify-generated-code
ci/prow/lint b5190b4 link true /test lint
ci/prow/e2e-metal-assisted-day2 b5190b4 link true /test e2e-metal-assisted-day2
ci/prow/e2e-metal-assisted-onprem b5190b4 link true /test e2e-metal-assisted-onprem
ci/prow/e2e-metal-assisted-static-ip-suite b5190b4 link true /test e2e-metal-assisted-static-ip-suite
ci/prow/e2e-metal-assisted-lvm b5190b4 link true /test e2e-metal-assisted-lvm

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@pastequo

Copy link
Copy Markdown
Contributor

Should be closed in favor of #6624

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

api-review Categorizes an issue or PR as actively needing an API review. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants