Skip to content

[cloud_hotfix_releases] MGMT-17620: Change the way multi-arch release images are handled in RestAPI flow - #6229

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:cloud_hotfix_releasesfrom
openshift-cherrypick-robot:cherry-pick-6227-to-cloud_hotfix_releases
Apr 21, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:cloud_hotfix_releasesfrom
openshift-cherrypick-robot:cherry-pick-6227-to-cloud_hotfix_releases

Conversation

@openshift-cherrypick-robot

Copy link
Copy Markdown

This is an automated cherry-pick of #6227

/assign danmanor

… to retrieve multiarch release image in case a multi-arch OpenShift version, single-arch requested and it exist.
@openshift-ci-robot

openshift-ci-robot commented Apr 21, 2024 •

Copy link
Copy Markdown

@openshift-cherrypick-robot: Ignoring requests to cherry-pick non-bug issues: MGMT-17620

Details

In response to this:

This is an automated cherry-pick of #6227

/assign danmanor

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Apr 21, 2024
@openshift-ci
openshift-ci Bot requested review from CrystalChun and eranco74 April 21, 2024 11:36
@codecov

codecov Bot commented Apr 21, 2024 •

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 68.22%. Comparing base (ef0c961) to head (9ff9b84).
Report is 993 commits behind head on cloud_hotfix_releases.

Additional details and impacted files

Impacted file tree graph

@@                    Coverage Diff                    @@
##           cloud_hotfix_releases    #6229      +/-   ##
=========================================================
+ Coverage                  66.66%   68.22%   +1.56%     
=========================================================
  Files                        197      242      +45     
  Lines                      27526    35861    +8335     
=========================================================
+ Hits                       18350    24467    +6117     
- Misses                      7531     9271    +1740     
- Partials                    1645     2123     +478     
Files Coverage Δ
internal/versions/rest_api_versions.go 71.00% <100.00%> (ø)

... and 159 files with indirect coverage changes

@danmanor

Copy link
Copy Markdown
Contributor

/approve

@danmanor

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 21, 2024
@openshift-ci

openshift-ci Bot commented Apr 21, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: danmanor, openshift-cherrypick-robot

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 21, 2024
@danmanor

Copy link
Copy Markdown
Contributor

/jira refresh

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Apr 21, 2024
@openshift-ci-robot

openshift-ci-robot commented Apr 21, 2024 •

Copy link
Copy Markdown

@danmanor: This pull request references MGMT-17620 which is a valid jira issue.

Details

In response to this:

/jira refresh

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci

openshift-ci Bot commented Apr 21, 2024

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit f022a08 into openshift:cloud_hotfix_releases Apr 21, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants