Skip to content

fix: presence for mobiles with duplicated connections - #31895

Merged
sampaiodiego merged 3 commits into
developfrom
fix-presence-duplicated-connections
Mar 5, 2024
Merged

fix: presence for mobiles with duplicated connections#31895
sampaiodiego merged 3 commits into
developfrom
fix-presence-duplicated-connections

Conversation

@sampaiodiego

@sampaiodiego sampaiodiego commented Mar 5, 2024

Copy link
Copy Markdown
Member

Proposed changes (including videos or screenshots)

The changes from #31487 added a connection ID for REST login requests, which caused an undesired effect of those requests being considered WS connections. That doesn't affect the web client since it only uses REST login requests with username and password (it doesn't use the "resume" login type). But the mobile uses the "resume" login type via REST to validate if stored credentials are still valid, which was causing "duplicated connection" entries being created for mobile, leading to mobile users being considered online forever (until a server restart).

With these changes we're making sure for every "resume" login action there is indeed an opened WS connection before we create this connection record on DB.

Issue(s)

closes #31768
closes #31873

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Mar 5, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7203870

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 31 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/apps Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Mar 5, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 54.60%. Comparing base (0f0d631) to head (7203870).

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31895      +/-   ##
===========================================
+ Coverage    54.56%   54.60%   +0.03%     
===========================================
  Files         2280     2283       +3     
  Lines        50244    50279      +35     
  Branches     10248    10255       +7     
===========================================
+ Hits         27418    27455      +37     
+ Misses       20340    20336       -4     
- Partials      2486     2488       +2     
Flag Coverage Δ
e2e 53.55% <ø> (+0.03%) ⬆️
e2e-api 40.09% <ø> (+0.06%) ⬆️
unit 76.44% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@sampaiodiego
sampaiodiego force-pushed the fix-presence-duplicated-connections branch from 0e4d715 to 7203870 Compare March 5, 2024 17:24
@sampaiodiego
sampaiodiego requested a review from a team as a code owner March 5, 2024 17:24
@sampaiodiego
sampaiodiego merged commit 7cb4f07 into develop Mar 5, 2024
@sampaiodiego
sampaiodiego deleted the fix-presence-duplicated-connections branch March 5, 2024 20:32
gabriellsh added a commit that referenced this pull request Mar 6, 2024
…tMarkdown

* 'develop' of github.com:RocketChat/Rocket.Chat:
  chore: improve useJumpToMessage (#31907)
  feat: Add `link` action to composer toolbar (#31679)
  chore: replace `PaletteStyleTag` in favor of fuselage component (#31839)
  fix: presence for mobiles with duplicated connections (#31895)
  fix: New users are not automatically inserted in auto-join rooms inside default teams (#31720)
  fix: livechat agent status set to available after reactivation (#31651)
  chore: Deprecate `insertOrUpdateUser` method (#31884)
  chore: update ts version on recently merged package (#31885)
  chore: move utilitary functions out of ee folder (#31207)
  chore: Remove references to EE code from the app bridges (#31757)
  fix: Room import doesn't honor the specified owner (#31803)
  fix: message parser emphasis black hole (#31868)
  chore: remove hardcoded position for message date (#31866)
  feat: show date on message's scroll (#31572)
  chore: `RoomHeader` keyboard navigability (#31837)
  fix: Login page breaking when handling not expected errors (#31804)
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

User presence issues in 6.6.2 Online status issue

3 participants