Skip to content

OCPBUGS-20097: Tests to rhel9 - #28734

Closed
joepvd wants to merge 1 commit into
openshift:masterfrom
joepvd:tests-to-rhel9
Closed

OCPBUGS-20097: Tests to rhel9#28734
joepvd wants to merge 1 commit into
openshift:masterfrom
joepvd:tests-to-rhel9

Conversation

@joepvd

@joepvd joepvd commented Apr 23, 2024

Copy link
Copy Markdown
Contributor

The package rt-tests is now called realtime-tests

The package rt-tests is now called realtime-tests

@soltysh soltysh left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@soltysh soltysh changed the title Tests to rhel9 OCPBUGS-20097: Tests to rhel9 Apr 23, 2024
@openshift-ci-robot openshift-ci-robot added jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. labels Apr 23, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@joepvd: This pull request references Jira Issue OCPBUGS-20097, which is valid.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.16.0) matches configured target version for branch (4.16.0)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

The package rt-tests is now called realtime-tests

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 23, 2024
@openshift-ci

openshift-ci Bot commented Apr 23, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: joepvd, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 23, 2024
@soltysh

soltysh commented Apr 23, 2024

Copy link
Copy Markdown
Contributor

Pulled that change into #28733

@soltysh soltysh closed this Apr 23, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@joepvd: This pull request references Jira Issue OCPBUGS-20097. The bug has been updated to no longer refer to the pull request using the external bug tracker.

Details

In response to this:

The package rt-tests is now called realtime-tests

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants