Skip to content

CORS-2750: Move to rhel9, decouple from cli image - #5792

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
joepvd:4.16-rhel9
May 1, 2024
Merged

CORS-2750: Move to rhel9, decouple from cli image#5792
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
joepvd:4.16-rhel9

Conversation

@joepvd

@joepvd joepvd commented Dec 9, 2023

Copy link
Copy Markdown
Contributor

List all the issues related to this PR

Move the image to rhel9. To do so, the openshift-clients rpm is installed rather than the oc binary plucked from the cli image.

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

@openshift-ci openshift-ci Bot added do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Dec 9, 2023
@openshift-ci
openshift-ci Bot requested review from danmanor and gamli75 December 9, 2023 11:37
@openshift-ci openshift-ci Bot added the kind/dependency-change Categorizes issue or PR as related to changing dependencies label Dec 9, 2023
@codecov

codecov Bot commented Dec 9, 2023

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 70.38%. Comparing base (9823c09) to head (4f6d88c).
Report is 63 commits behind head on master.

Additional details and impacted files

Impacted file tree graph

@@            Coverage Diff             @@
##           master    #5792      +/-   ##
==========================================
+ Coverage   68.40%   70.38%   +1.97%     
==========================================
  Files         239      246       +7     
  Lines       35457    41244    +5787     
==========================================
+ Hits        24256    29030    +4774     
- Misses       9098     9787     +689     
- Partials     2103     2427     +324     

see 38 files with indirect coverage changes

@joepvd joepvd changed the title wip: Move to rhel9, decouple from cli image OCPBUGS-25037: Move to rhel9, decouple from cli image Dec 11, 2023
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Dec 11, 2023
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Dec 11, 2023
@openshift-ci-robot

Copy link
Copy Markdown

@joepvd: This pull request references Jira Issue OCPBUGS-25037, which is valid. The bug has been moved to the POST state.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.16.0) matches configured target version for branch (4.16.0)
  • bug is in the state New, which is one of the valid states (NEW, ASSIGNED, POST)

No GitHub users were found matching the public email listed for the QA contact in Jira (lalon@redhat.com), skipping review request.

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

List all the issues related to this PR

Move the image to rhel9. To do so, the openshift-clients rpm is installed rather than the oc binary plucked from the cli image.

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot openshift-ci-robot added the jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. label Dec 11, 2023
@joepvd

joepvd commented Dec 11, 2023

Copy link
Copy Markdown
Contributor Author

/retest

1 similar comment
@adriengentil

Copy link
Copy Markdown
Contributor

/retest

@joepvd

joepvd commented Dec 12, 2023

Copy link
Copy Markdown
Contributor Author

/hold
brew is not ready yet to build on rhel9

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Dec 12, 2023
@joepvd

joepvd commented Mar 11, 2024

Copy link
Copy Markdown
Contributor Author

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Mar 11, 2024
@adriengentil

Copy link
Copy Markdown
Contributor

/retest
/cc @andfasano

@openshift-ci
openshift-ci Bot requested a review from andfasano March 11, 2024 13:23
@zaneb

zaneb commented Mar 11, 2024

Copy link
Copy Markdown
Member

For the e2e-agent-compact-ipv4 build (which has FIPS enabled) to pass, we need the assisted-service container to also be RHEL9-based.
Of course changing that one first will also fail. Since we can't make the change atomic we'll have to have both PRs ready to go, override the tests, and merge both as close together as possible.

@zaneb

zaneb commented Apr 16, 2024

Copy link
Copy Markdown
Member

/retitle CORS-2750: Move to rhel9, decouple from cli image

@openshift-ci openshift-ci Bot changed the title OCPBUGS-25037: Move to rhel9, decouple from cli image CORS-2750: Move to rhel9, decouple from cli image Apr 16, 2024
@openshift-ci-robot openshift-ci-robot removed the jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. label Apr 16, 2024
@openshift-ci-robot

openshift-ci-robot commented Apr 16, 2024

Copy link
Copy Markdown

@joepvd: This pull request references CORS-2750 which is a valid jira issue.

Details

In response to this:

List all the issues related to this PR

Move the image to rhel9. To do so, the openshift-clients rpm is installed rather than the oc binary plucked from the cli image.

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@joepvd

joepvd commented Apr 23, 2024

Copy link
Copy Markdown
Contributor Author

/hold
Needs to go together with openshift/installer#8196

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 23, 2024
@carbonin

Copy link
Copy Markdown
Member

/lgtm

@carbonin

Copy link
Copy Markdown
Member

/override ci/prow/e2e-agent-compact-ipv4

@carbonin

Copy link
Copy Markdown
Member

/approve

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 23, 2024
@openshift-ci

openshift-ci Bot commented Apr 23, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: carbonin, joepvd

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 23, 2024
@openshift-ci

openshift-ci Bot commented Apr 23, 2024

Copy link
Copy Markdown
Contributor

@carbonin: Overrode contexts on behalf of carbonin: ci/prow/e2e-agent-compact-ipv4

Details

In response to this:

/override ci/prow/e2e-agent-compact-ipv4

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@thegreyd

Copy link
Copy Markdown

/unhold

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 30, 2024
@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 8c73c0e and 2 for PR HEAD 4f6d88c in total

@zaneb

zaneb commented Apr 30, 2024

Copy link
Copy Markdown
Member

/retest-required

@zaneb

zaneb commented May 1, 2024

Copy link
Copy Markdown
Member

edge-e2e-metal-assisted tests are failing with:

[2024-04-30 22:16:51] openshift-tests: /lib64/libc.so.6: version `GLIBC_2.34'\'' not found (required by openshift-tests)
[2024-04-30 22:16:51] openshift-tests: /lib64/libc.so.6: version `GLIBC_2.32'\'' not found (required by openshift-tests)'

but it is not due to this PR (other PRs have the same error). Presumably it is caused by openshift/origin#28733

@zaneb

zaneb commented May 1, 2024

Copy link
Copy Markdown
Member

I don't think I can override, but I'll give it a go because this is blocking openshift-eng/ocp-build-data#4659 which is blocking the agent release informing job.
/override ci/prow/edge-e2e-metal-assisted

@openshift-ci

openshift-ci Bot commented May 1, 2024

Copy link
Copy Markdown
Contributor

@zaneb: Overrode contexts on behalf of zaneb: ci/prow/edge-e2e-metal-assisted

Details

In response to this:

I don't think I can override, but I'll give it a go because this is blocking openshift-eng/ocp-build-data#4659 which is blocking the agent release informing job.
/override ci/prow/edge-e2e-metal-assisted

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci

openshift-ci Bot commented May 1, 2024

Copy link
Copy Markdown
Contributor

@joepvd: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 6a70f70 into openshift:master May 1, 2024
@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

This PR has been included in build ose-agent-installer-api-server-container-v4.17.0-202405010518.p0.g6a70f70.assembly.stream.el9 for distgit ose-agent-installer-api-server.
All builds following this will include this PR.

@adriengentil adriengentil mentioned this pull request Jun 5, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. kind/dependency-change Categorizes issue or PR as related to changing dependencies lgtm Indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants