Skip to content

Revert "Fix tests" - #4719

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift-4.17from
revert-4717-try-tests
Apr 24, 2024
Merged

Revert "Fix tests"#4719
openshift-merge-bot[bot] merged 1 commit into
openshift-4.17from
revert-4717-try-tests

Conversation

@joepvd

@joepvd joepvd commented Apr 24, 2024

Copy link
Copy Markdown
Contributor

Reverts #4717

To be merged when openshift/origin#28733 lands

@joepvd

joepvd commented Apr 24, 2024

Copy link
Copy Markdown
Contributor Author

/hold

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 24, 2024

@ashwindasr ashwindasr left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/unhold

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 24, 2024
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 24, 2024
@openshift-ci

openshift-ci Bot commented Apr 24, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ashwindasr

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 24, 2024
@openshift-merge-bot
openshift-merge-bot Bot merged commit efb928b into openshift-4.17 Apr 24, 2024
@openshift-merge-bot
openshift-merge-bot Bot deleted the revert-4717-try-tests branch April 24, 2024 17:32
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants