Skip to content

NO-JIRA: Drop unused images and OWNERS files - #28736

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
soltysh:clean_images
Apr 24, 2024
Merged

NO-JIRA: Drop unused images and OWNERS files#28736
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
soltysh:clean_images

Conversation

@soltysh

@soltysh soltysh commented Apr 24, 2024

Copy link
Copy Markdown
Contributor

No description provided.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Apr 24, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@soltysh: This pull request explicitly references no jira issue.

Details

In response to this:

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@ardaguclu

Copy link
Copy Markdown
Member

/lgtm

@soltysh

soltysh commented Apr 24, 2024

Copy link
Copy Markdown
Contributor Author

/hold
until we land #28733

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 24, 2024
@openshift-ci
openshift-ci Bot requested review from knobunc and sdodson April 24, 2024 09:35
@openshift-ci

openshift-ci Bot commented Apr 24, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ardaguclu, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 24, 2024
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 24, 2024
@soltysh

soltysh commented Apr 24, 2024

Copy link
Copy Markdown
Contributor Author

/retest

@soltysh

soltysh commented Apr 24, 2024

Copy link
Copy Markdown
Contributor Author

/test images

@soltysh

soltysh commented Apr 24, 2024

Copy link
Copy Markdown
Contributor Author

/retest-required

@soltysh

soltysh commented Apr 24, 2024

Copy link
Copy Markdown
Contributor Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 24, 2024
@openshift-trt-bot

Copy link
Copy Markdown

Job Failure Risk Analysis for sha: 51b1dbf

Job Name Failure Risk
pull-ci-openshift-origin-master-e2e-openstack-ovn IncompleteTests
Tests for this run (14) are below the historical average (1541): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-metal-ipi-sdn IncompleteTests
Tests for this run (13) are below the historical average (971): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-gcp-ovn-rt-upgrade IncompleteTests
Tests for this run (17) are below the historical average (568): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-gcp-csi IncompleteTests
Tests for this run (16) are below the historical average (552): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-aws-ovn-upgrade IncompleteTests
Tests for this run (18) are below the historical average (657): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-aws-ovn-single-node-upgrade IncompleteTests
Tests for this run (17) are below the historical average (2076): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-aws-ovn-single-node-serial IncompleteTests
Tests for this run (16) are below the historical average (699): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-aws-ovn-single-node IncompleteTests
Tests for this run (16) are below the historical average (1552): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-aws-ovn-cgroupsv2 IncompleteTests
Tests for this run (16) are below the historical average (1555): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-aws-csi IncompleteTests
Tests for this run (16) are below the historical average (653): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-agnostic-ovn-cmd IncompleteTests
Tests for this run (16) are below the historical average (518): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)

@openshift-ci

openshift-ci Bot commented Apr 24, 2024

Copy link
Copy Markdown
Contributor

@soltysh: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-gcp-csi 51b1dbf link false /test e2e-gcp-csi
ci/prow/e2e-aws-ovn-single-node 51b1dbf link false /test e2e-aws-ovn-single-node
ci/prow/e2e-metal-ipi-sdn 51b1dbf link false /test e2e-metal-ipi-sdn
ci/prow/e2e-agnostic-ovn-cmd 51b1dbf link false /test e2e-agnostic-ovn-cmd
ci/prow/e2e-gcp-ovn-rt-upgrade 51b1dbf link false /test e2e-gcp-ovn-rt-upgrade
ci/prow/e2e-openstack-ovn 51b1dbf link false /test e2e-openstack-ovn
ci/prow/e2e-aws-ovn-cgroupsv2 51b1dbf link false /test e2e-aws-ovn-cgroupsv2
ci/prow/e2e-aws-csi 51b1dbf link false /test e2e-aws-csi
ci/prow/e2e-aws-ovn-single-node-serial 51b1dbf link false /test e2e-aws-ovn-single-node-serial
ci/prow/e2e-aws-ovn-single-node-upgrade 51b1dbf link false /test e2e-aws-ovn-single-node-upgrade
ci/prow/e2e-aws-ovn-upgrade 51b1dbf link false /test e2e-aws-ovn-upgrade

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit cab2b73 into openshift:master Apr 24, 2024
@soltysh
soltysh deleted the clean_images branch April 25, 2024 09:03
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants