feat(uikit): Move @rocket.chat/ui-kit package to the main monorepo - #31138
Conversation
🦋 Changeset detectedLatest commit: bdb1f68 The changes in this PR will be included in the next version bump. This PR includes changesets to release 52 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov Report
Additional details and impacted files@@ Coverage Diff @@
## develop #31138 +/- ##
===========================================
- Coverage 57.24% 48.73% -8.51%
===========================================
Files 881 3280 +2399
Lines 19949 80006 +60057
Branches 3719 16011 +12292
===========================================
+ Hits 11419 38992 +27573
- Misses 7841 36438 +28597
- Partials 689 4576 +3887
Flags with carried forward coverage won't be shown. Click here to find out more. |
bb5741f to
1742952
Compare
6f91653 to
70f190d
Compare
| "@rocket.chat/ui-composer": "workspace:^", | ||
| "@rocket.chat/ui-contexts": "workspace:^", | ||
| "@rocket.chat/ui-kit": "~0.32.1", | ||
| "@rocket.chat/ui-kit": "workspace:~", |
There was a problem hiding this comment.
any special reason to use workspace:~ instead of workspace:^? this used to be an issue with the script to fix the version dependencies before publishing a package, but now we just ignore whatever "symbol" is being used..
so my question is more why did you chose for ~? since we use ^ for everything else
There was a problem hiding this comment.
I particularly go for yarn add --tilde when adding new dependency because "tilde-dependencies" deny minor-version bumps, while carets not; in my personal projects, I even put defaultSemverRangePrefix: "~" on .yarnrc.yml.
When the package is published on npmjs registry, workspace:~ is replaced with ~x.y.z while workspace:^ is replaced with ^x.y.z.
There was a problem hiding this comment.
When the package is published on npmjs registry,
workspace:~is replaced with~x.y.zwhileworkspace:^is replaced with^x.y.z.
not really, we always set to ^x.y.z :
There was a problem hiding this comment.
Ideally we shouldn't have to do this heavy-lifting ourselves... https://yarnpkg.com/features/workspaces#cross-references
There was a problem hiding this comment.
"Na prática, a teoria é outra."
There was a problem hiding this comment.
Patched the action's code.
…hideUi * 'develop' of github.com:RocketChat/Rocket.Chat: (21 commits) chore: Rename useFileInput to useSingleFileInput (#31186) regression: FED-116 allow to invite rc users to fed channels (#30658) refactor: Remove `livechat.js` file (#31179) test: Fix Analytics test to use always same date (#31178) chore: notify messages on emails incoming and outgoing (#30925) chore: add msg stream to the federation v1 feature (#31105) chore: update editedBy when the user changes their username (following BE rules) (#31098) chore: add a manual msg stream update for autotranslate (#31094) chore: send streaming msg update on livechat decline call event (#31025) chore: Replace `IconButton` in favor of `Button` on `PageHeader` (#31171) refactor: Migrate `analytics.js` to service (#31051) feat: Add `push.info` endpoint to retrieve push gateway info (#30868) chore: Convert Importer classes to typescript (#29714) docs: fix lost image render in markdown (#31013) feat(uikit): Move `@rocket.chat/ui-kit` package to the main monorepo (#31138) i18n: Language update from LingoHub 🤖 on 2023-12-05Z (#31161) feat: Skip to main content shortcut and `useDocumentTitle` (#30680) chore: Replace `useForm` in favor of RHF on Omnichannel `AgentsEdit` (#30789) chore: ImageGallery tweaks (#31159) fix: New `custom-roles` license module isn't properly checked (#31153) ...
Proposed changes (including videos or screenshots)
Issue(s)
Steps to test or reproduce
Further comments