Skip to content

regression: FED-116 allow to invite rc users to fed channels - #30658

Merged
ggazzo merged 13 commits into
developfrom
fix/fed-116
Dec 7, 2023
Merged

regression: FED-116 allow to invite rc users to fed channels#30658
ggazzo merged 13 commits into
developfrom
fix/fed-116

Conversation

@lmauromb

@lmauromb lmauromb commented Oct 17, 2023

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

Fix regression when inviting users on fed channels
Creates FederationServiceEE correctly

Other:

Issue(s)

https://rocketchat.atlassian.net/browse/FED-116

@lmauromb
lmauromb requested a review from a team as a code owner October 17, 2023 16:52
@changeset-bot

changeset-bot Bot commented Oct 17, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 30c222f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/rest-typings Minor
@rocket.chat/meteor Minor
@rocket.chat/core-services Patch
@rocket.chat/ui-contexts Major
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/ddp-client Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/fuselage-ui-kit Major
@rocket.chat/gazzodown Major
@rocket.chat/ui-client Major
@rocket.chat/ui-video-conf Major
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Major
@rocket.chat/livechat Patch
@rocket.chat/core-typings Minor
@rocket.chat/cron Patch
@rocket.chat/model-typings Patch
@rocket.chat/license Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/models Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@lmauromb
lmauromb requested a review from ggazzo October 17, 2023 16:52
@codecov

codecov Bot commented Oct 17, 2023

Copy link
Copy Markdown

Codecov Report

Merging #30658 (30c222f) into develop (b1614cb) will increase coverage by 4.72%.
The diff coverage is n/a.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #30658      +/-   ##
===========================================
+ Coverage    46.31%   51.04%   +4.72%     
===========================================
  Files          766      807      +41     
  Lines        14675    15141     +466     
  Branches      2753     2820      +67     
===========================================
+ Hits          6797     7728     +931     
+ Misses        7479     6977     -502     
- Partials       399      436      +37     
Flag Coverage Δ
e2e 47.93% <ø> (+5.52%) ⬆️
unit 66.21% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@lmauromb
lmauromb requested review from a team as code owners October 27, 2023 01:31
Comment thread packages/rest-typings/src/v1/federation/FederationVerifyMatrixIdProps.ts Outdated
@lmauromb
lmauromb requested review from a team as code owners October 30, 2023 17:01
@lmauromb
lmauromb removed request for a team October 30, 2023 17:14
MarcosSpessatto
MarcosSpessatto previously approved these changes Nov 1, 2023
@lmauromb lmauromb changed the title fix: FED-116 allow to invite rc users to fed channels regression: FED-116 allow to invite rc users to fed channels Nov 3, 2023
@ggazzo
ggazzo merged commit a271587 into develop Dec 7, 2023
@ggazzo
ggazzo deleted the fix/fed-116 branch December 7, 2023 16:08
gabriellsh added a commit that referenced this pull request Dec 7, 2023
…hideUi

* 'develop' of github.com:RocketChat/Rocket.Chat: (21 commits)
  chore: Rename useFileInput to useSingleFileInput (#31186)
  regression: FED-116 allow to invite rc users to fed channels (#30658)
  refactor: Remove `livechat.js` file (#31179)
  test: Fix Analytics test to use always same date (#31178)
  chore: notify messages on emails incoming and outgoing (#30925)
  chore: add msg stream to the federation v1 feature (#31105)
  chore: update editedBy when the user changes their username (following BE rules) (#31098)
  chore: add a manual msg stream update for autotranslate (#31094)
  chore: send streaming msg update on livechat decline call event (#31025)
  chore: Replace `IconButton` in favor of `Button` on `PageHeader` (#31171)
  refactor: Migrate `analytics.js` to service (#31051)
  feat: Add `push.info` endpoint to retrieve push gateway info (#30868)
  chore: Convert Importer classes to typescript (#29714)
  docs: fix lost image render in markdown (#31013)
  feat(uikit): Move `@rocket.chat/ui-kit` package to the main monorepo (#31138)
  i18n: Language update from LingoHub 🤖 on 2023-12-05Z (#31161)
  feat: Skip to main content shortcut and `useDocumentTitle` (#30680)
  chore: Replace `useForm` in favor of RHF on Omnichannel `AgentsEdit` (#30789)
  chore: ImageGallery tweaks (#31159)
  fix: New `custom-roles` license module isn't properly checked (#31153)
  ...
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants