Skip to content

feat: Skip to main content shortcut and useDocumentTitle - #30680

Merged
ggazzo merged 19 commits into
developfrom
chore/documentTitleWrapper
Dec 5, 2023
Merged

feat: Skip to main content shortcut and useDocumentTitle#30680
ggazzo merged 19 commits into
developfrom
chore/documentTitleWrapper

Conversation

@dougfabris

@dougfabris dougfabris commented Oct 18, 2023

Copy link
Copy Markdown
Member

Proposed changes (including videos or screenshots)

  • Adds a shortcut to skip the focus to the main content when navigating through the keyboard
  • useDocumentTitle to change the document title dynamically

Kapture 2023-12-05 at 15 02 00

Issue(s)

Steps to test or reproduce

Further comments

DSAR-226
DSAR-286
DSAR-287

@changeset-bot

changeset-bot Bot commented Oct 18, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b97ed34

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Minor
@rocket.chat/ui-client Major
@rocket.chat/web-ui-registration Major
@rocket.chat/gazzodown Major
@rocket.chat/fuselage-ui-kit Major
@rocket.chat/livechat Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/core-typings Minor
@rocket.chat/rest-typings Minor
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Major
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/models Patch
@rocket.chat/ui-video-conf Major
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Oct 18, 2023

Copy link
Copy Markdown

Codecov Report

Merging #30680 (b97ed34) into develop (b90f871) will decrease coverage by 8.37%.
Report is 6 commits behind head on develop.
The diff coverage is 92.18%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #30680      +/-   ##
===========================================
- Coverage    57.14%   48.77%   -8.37%     
===========================================
  Files          880     3283    +2403     
  Lines        19941    80045   +60104     
  Branches      3719    16017   +12298     
===========================================
+ Hits         11395    39044   +27649     
- Misses        7857    36421   +28564     
- Partials       689     4580    +3891     
Flag Coverage Δ
e2e 51.51% <86.84%> (+9.18%) ⬆️
e2e-api 40.76% <ø> (?)
unit 76.70% <100.00%> (+0.07%) ⬆️

Flags with carried forward coverage won't be shown. Click here to find out more.

@dougfabris
dougfabris marked this pull request as ready for review December 5, 2023 14:50
@dougfabris
dougfabris requested a review from a team as a code owner December 5, 2023 14:50
@dougfabris dougfabris changed the title chore: Change document title based on routes feat: Change document title based on routes Dec 5, 2023
@dougfabris dougfabris changed the title feat: Change document title based on routes feat: Skip to main content shortcut and useDocumentTitle Dec 5, 2023
@ggazzo
ggazzo merged commit dd5fd6d into develop Dec 5, 2023
@ggazzo
ggazzo deleted the chore/documentTitleWrapper branch December 5, 2023 19:40
gabriellsh added a commit that referenced this pull request Dec 7, 2023
…hideUi

* 'develop' of github.com:RocketChat/Rocket.Chat: (21 commits)
  chore: Rename useFileInput to useSingleFileInput (#31186)
  regression: FED-116 allow to invite rc users to fed channels (#30658)
  refactor: Remove `livechat.js` file (#31179)
  test: Fix Analytics test to use always same date (#31178)
  chore: notify messages on emails incoming and outgoing (#30925)
  chore: add msg stream to the federation v1 feature (#31105)
  chore: update editedBy when the user changes their username (following BE rules) (#31098)
  chore: add a manual msg stream update for autotranslate (#31094)
  chore: send streaming msg update on livechat decline call event (#31025)
  chore: Replace `IconButton` in favor of `Button` on `PageHeader` (#31171)
  refactor: Migrate `analytics.js` to service (#31051)
  feat: Add `push.info` endpoint to retrieve push gateway info (#30868)
  chore: Convert Importer classes to typescript (#29714)
  docs: fix lost image render in markdown (#31013)
  feat(uikit): Move `@rocket.chat/ui-kit` package to the main monorepo (#31138)
  i18n: Language update from LingoHub 🤖 on 2023-12-05Z (#31161)
  feat: Skip to main content shortcut and `useDocumentTitle` (#30680)
  chore: Replace `useForm` in favor of RHF on Omnichannel `AgentsEdit` (#30789)
  chore: ImageGallery tweaks (#31159)
  fix: New `custom-roles` license module isn't properly checked (#31153)
  ...
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants