Skip to content

Release 6.6.3 - #31903

Merged
sampaiodiego merged 2 commits into
masterfrom
release-6.6.3
Mar 6, 2024
Merged

Release 6.6.3#31903
sampaiodiego merged 2 commits into
masterfrom
release-6.6.3

Conversation

@rocketchat-github-ci

@rocketchat-github-ci rocketchat-github-ci commented Mar 5, 2024

Copy link
Copy Markdown
Collaborator

You can see below a preview of the release change log:

6.6.3

Engine versions

  • Node: 14.21.3
  • MongoDB: 4.4, 5.0, 6.0
  • Apps-Engine: 1.41.0

Patch Changes

  • Bump @rocket.chat/meteor version.

  • (#31895) Fix users presence stuck as online after connecting using mobile apps

  • Updated dependencies []:
    • @rocket.chat/core-typings@6.6.3
    • @rocket.chat/rest-typings@6.6.3
    • @rocket.chat/api-client@0.1.25
    • @rocket.chat/license@0.1.7
    • @rocket.chat/omnichannel-services@0.1.7
    • @rocket.chat/pdf-worker@0.0.31
    • @rocket.chat/presence@0.1.7
    • @rocket.chat/core-services@0.3.7
    • @rocket.chat/cron@0.0.27
    • @rocket.chat/gazzodown@4.0.3
    • @rocket.chat/model-typings@0.3.3
    • @rocket.chat/ui-contexts@4.0.3
    • @rocket.chat/server-cloud-communication@0.0.2
    • @rocket.chat/fuselage-ui-kit@4.0.3
    • @rocket.chat/models@0.0.31
    • @rocket.chat/ui-theming@0.1.2
    • @rocket.chat/ui-client@4.0.3
    • @rocket.chat/ui-video-conf@4.0.3
    • @rocket.chat/web-ui-registration@4.0.3
    • @rocket.chat/instance-status@0.0.31

@changeset-bot

changeset-bot Bot commented Mar 5, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0398ecc

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Mar 5, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 54.52%. Comparing base (44b682f) to head (0398ecc).

Additional details and impacted files

Impacted file tree graph

@@            Coverage Diff             @@
##           master   #31903      +/-   ##
==========================================
+ Coverage   54.37%   54.52%   +0.14%     
==========================================
  Files        2267     2274       +7     
  Lines       49949    50040      +91     
  Branches    10200    10218      +18     
==========================================
+ Hits        27161    27284     +123     
+ Misses      20352    20274      -78     
- Partials     2436     2482      +46     
Flag Coverage Δ
e2e 53.37% <ø> (+0.17%) ⬆️
e2e-api 40.12% <ø> (+0.19%) ⬆️
unit 76.47% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@rocketchat-github-ci
rocketchat-github-ci requested a review from a team as a code owner March 5, 2024 20:34
@sampaiodiego
sampaiodiego merged commit 0eceb73 into master Mar 6, 2024
@sampaiodiego
sampaiodiego deleted the release-6.6.3 branch March 6, 2024 13:52
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants