Skip to content

chore: Deprecate insertOrUpdateUser method - #31884

Merged
kodiakhq[bot] merged 4 commits into
developfrom
chore/deprecate-insertOrUpdateUser
Mar 5, 2024
Merged

chore: Deprecate insertOrUpdateUser method#31884
kodiakhq[bot] merged 4 commits into
developfrom
chore/deprecate-insertOrUpdateUser

Conversation

@matheusbsilva137

@matheusbsilva137 matheusbsilva137 commented Mar 4, 2024

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

  • Deprecate insertOrUpdateUser Meteor method (planned to be removed on 8.0.0) -- this is part of an initiative for not allowing users upsert on RC.

Issue(s)

Steps to test or reproduce

Further comments

CORE-132

@changeset-bot

changeset-bot Bot commented Mar 4, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ad745d5

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 32 packages
Name Type
@rocket.chat/meteor Minor
@rocket.chat/core-typings Minor
@rocket.chat/rest-typings Minor
@rocket.chat/apps Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Major
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Major
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Major
@rocket.chat/models Patch
@rocket.chat/ui-avatar Major
@rocket.chat/ui-client Major
@rocket.chat/ui-video-conf Major
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Major
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@matheusbsilva137
matheusbsilva137 marked this pull request as ready for review March 4, 2024 18:41
@scuciatto scuciatto modified the milestones: 6.7, 7.0 Mar 5, 2024
@codecov

codecov Bot commented Mar 5, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 54.57%. Comparing base (cac5c26) to head (ad745d5).

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31884      +/-   ##
===========================================
+ Coverage    54.55%   54.57%   +0.01%     
===========================================
  Files         2283     2282       -1     
  Lines        50279    50248      -31     
  Branches     10255    10249       -6     
===========================================
- Hits         27430    27423       -7     
+ Misses       20358    20336      -22     
+ Partials      2491     2489       -2     
Flag Coverage Δ
e2e 53.56% <ø> (+0.03%) ⬆️
e2e-api 39.98% <ø> (+<0.01%) ⬆️
unit 76.44% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Mar 5, 2024
@kodiakhq
kodiakhq Bot merged commit a011734 into develop Mar 5, 2024
@kodiakhq
kodiakhq Bot deleted the chore/deprecate-insertOrUpdateUser branch March 5, 2024 12:41
gabriellsh added a commit that referenced this pull request Mar 6, 2024
…tMarkdown

* 'develop' of github.com:RocketChat/Rocket.Chat:
  chore: improve useJumpToMessage (#31907)
  feat: Add `link` action to composer toolbar (#31679)
  chore: replace `PaletteStyleTag` in favor of fuselage component (#31839)
  fix: presence for mobiles with duplicated connections (#31895)
  fix: New users are not automatically inserted in auto-join rooms inside default teams (#31720)
  fix: livechat agent status set to available after reactivation (#31651)
  chore: Deprecate `insertOrUpdateUser` method (#31884)
  chore: update ts version on recently merged package (#31885)
  chore: move utilitary functions out of ee folder (#31207)
  chore: Remove references to EE code from the app bridges (#31757)
  fix: Room import doesn't honor the specified owner (#31803)
  fix: message parser emphasis black hole (#31868)
  chore: remove hardcoded position for message date (#31866)
  feat: show date on message's scroll (#31572)
  chore: `RoomHeader` keyboard navigability (#31837)
  fix: Login page breaking when handling not expected errors (#31804)
@scuciatto scuciatto modified the milestones: 7.0, 6.7 Mar 6, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants