feat: added modal confirmation before pinning message - #31348
Conversation
🦋 Changeset detectedLatest commit: 327b50f The changes in this PR will be included in the next version bump. This PR includes changesets to release 30 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov ReportAttention:
Additional details and impacted files@@ Coverage Diff @@
## develop #31348 +/- ##
============================================
- Coverage 59.38% 49.36% -10.03%
============================================
Files 1793 3311 +1518
Lines 34909 81366 +46457
Branches 7246 16661 +9415
============================================
+ Hits 20731 40163 +19432
- Misses 12577 36501 +23924
- Partials 1601 4702 +3101
Flags with carried forward coverage won't be shown. Click here to find out more. |
There was a problem hiding this comment.
@Sayan4444 This is the desired result shared by our design team, please check if you can achieve it using <GenericModal />, if you have any questions, let me know!
Also, can you check what happened with the en.i18n.json?
|
@dougfabris Is this good enough or do you want me to add or correct anything else?? Can you tell me what is wrong about en.i18n.json?, it will be of greate help |
d7121f4 to
de44571
Compare
There was a problem hiding this comment.
@Sayan4444 Thank you so much for the contribution, it was very precise and valuable.
About the en.i18n.json file I think it was some issue syncing your code or with lint.
I made some small changes, just to match our standards and code organization.
Thanks again and I hope to see you contributing more 🚀
|
@dougfabris Yep I will be contributing more as much as I can as I love solving issues which directly impact the end user experience. Do you have an issue which u are also waiting long time to see someone working on it? |
|
@Sayan4444 That's awesome to hear from you. |
…hideUi * 'develop' of github.com:RocketChat/Rocket.Chat: chore: add gitpod config file for quick setup (#30921) feat: Allow user to set timeout of outgoing webhook (#31222) feat: added modal confirmation before pinning message (#31348) fix: not being able to access or create rooms with join code (#31270) chore: services changes on lifecycle methods + throw if svc is not available (#31375) test: bump playwright (#31376) i18n: update translations (#29462) i18n: adds video call translations for Persian language (#30406) regression: `AppRow` bundleIn verification (#31373) feat: Add Desktop PDF viewer (#31279) fix: Attachments not collapsing when using incoming webhook (#31318) chore: add aria-label to Select Inputs at Engagement Dashboard (#31249) fix: engagement dashboard timezone selector (#31248)

Proposed changes (including videos or screenshots)
AFTER THE FIX
https://github.com/RocketChat/Rocket.Chat/assets/112304873/2e74a44f-e168-48af-817d-c7b945d1f5b0
REASON BEHIND CHANGES
Made the same changes as done by DISCORD to tackle this issue.
Issue(s)
This should close issues :-
closes #30777
Steps to test or reproduce