Skip to content

fix: engagement dashboard timezone selector - #31248

Merged
kodiakhq[bot] merged 10 commits into
developfrom
fix/engagement-dashboard-timezone-selector
Jan 3, 2024
Merged

fix: engagement dashboard timezone selector#31248
kodiakhq[bot] merged 10 commits into
developfrom
fix/engagement-dashboard-timezone-selector

Conversation

@hugocostadev

@hugocostadev hugocostadev commented Dec 15, 2023

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

Due to the ResponsiveHeatMap component work with SVG and it edits a lot's the DOM tree in every Timezone changes, the UI gets frozen for a bit.

Upgrading the nivo package and replacing the ResponsiveHeatMap with ResponsiveHeatMapCanvas fixed the issue.

Issue(s)

Steps to test or reproduce

Further comments

https://rocketchat.atlassian.net/browse/WM-198

@hugocostadev hugocostadev added this to the 6.6 milestone Dec 15, 2023
@changeset-bot

changeset-bot Bot commented Dec 15, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f8337a1

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Dec 15, 2023

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Comparison is base (adbc874) 76.67% compared to head (f8337a1) 59.41%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff              @@
##           develop   #31248       +/-   ##
============================================
- Coverage    76.67%   59.41%   -17.27%     
============================================
  Files          277     1793     +1516     
  Lines         8966    34900    +25934     
  Branches      1695     7244     +5549     
============================================
+ Hits          6875    20735    +13860     
- Misses        1667    12567    +10900     
- Partials       424     1598     +1174     
Flag Coverage Δ
e2e 52.60% <ø> (?)
unit 76.67% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@casalsgh
casalsgh requested a review from a team December 26, 2023 16:57
dougfabris
dougfabris previously approved these changes Dec 27, 2023

@dougfabris dougfabris left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM!

@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Jan 2, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 2, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 3, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 3, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 3, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 3, 2024
@kodiakhq
kodiakhq Bot merged commit 18a9d65 into develop Jan 3, 2024
@kodiakhq
kodiakhq Bot deleted the fix/engagement-dashboard-timezone-selector branch January 3, 2024 15:28
gabriellsh added a commit that referenced this pull request Jan 5, 2024
…hideUi

* 'develop' of github.com:RocketChat/Rocket.Chat:
  chore: add gitpod config file for quick setup (#30921)
  feat: Allow user to set timeout of outgoing webhook (#31222)
  feat: added modal confirmation before pinning message (#31348)
  fix: not being able to access or create rooms with join code (#31270)
  chore: services changes on lifecycle methods + throw if svc is not available (#31375)
  test: bump playwright (#31376)
  i18n: update translations (#29462)
  i18n: adds video call translations for Persian language (#30406)
  regression: `AppRow` bundleIn verification (#31373)
  feat: Add Desktop PDF viewer (#31279)
  fix: Attachments not collapsing when using incoming webhook (#31318)
  chore: add aria-label to Select Inputs at Engagement Dashboard (#31249)
  fix: engagement dashboard timezone selector (#31248)
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants