Skip to content

OCPBUGS-32293: Update Jenkins test to use Deployment - #28718

Closed
apoorvajagtap wants to merge 2 commits into
openshift:masterfrom
apoorvajagtap:OCPBUGS-32293
Closed

OCPBUGS-32293: Update Jenkins test to use Deployment#28718
apoorvajagtap wants to merge 2 commits into
openshift:masterfrom
apoorvajagtap:OCPBUGS-32293

Conversation

@apoorvajagtap

Copy link
Copy Markdown
Contributor

Failing Jenkins Pipeline test blocks multiple PRs from merging. As agreed, skipping these tests temporarily.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Apr 17, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@apoorvajagtap: This pull request references Jira Issue OCPBUGS-32293, which is valid. The bug has been moved to the POST state.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.16.0) matches configured target version for branch (4.16.0)
  • bug is in the state ASSIGNED, which is one of the valid states (NEW, ASSIGNED, POST)

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

Failing Jenkins Pipeline test blocks multiple PRs from merging. As agreed, skipping these tests temporarily.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot openshift-ci-robot added the jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. label Apr 17, 2024
@apoorvajagtap

Copy link
Copy Markdown
Contributor Author

/assign @adambkaplan

@adambkaplan adambkaplan left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve

Minor suggestion on the skip message.

Comment thread test/extended/builds/pipeline_origin_bld.go Outdated
@sayan-biswas

Copy link
Copy Markdown

/approve

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 17, 2024
@sayan-biswas

Copy link
Copy Markdown

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 18, 2024
@openshift-ci

openshift-ci Bot commented Apr 18, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: adambkaplan, apoorvajagtap, sayan-biswas

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 886b7ae and 2 for PR HEAD 809e832 in total

@openshift-trt-bot

Copy link
Copy Markdown

Job Failure Risk Analysis for sha: 809e832

Job Name Failure Risk
pull-ci-openshift-origin-master-e2e-metal-ipi-ovn-ipv6 IncompleteTests
Tests for this run (18) are below the historical average (741): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)

@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Apr 18, 2024
@openshift-ci

openshift-ci Bot commented Apr 18, 2024

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@apoorvajagtap

Copy link
Copy Markdown
Contributor Author

@adambkaplan While looking through the test logs, I realized that it creates deployment (template changes introduced with 4.15) & the test waits for deploymentConfig to be ready.
Instead of skipping, I have replaced the wait function, and I hope this will fix it. PTAL :)

@adambkaplan adambkaplan changed the title OCPBUGS-32293: Skips Jenkins pipeline build config strategy test temporarily OCPBUGS-32293: Update Jenkins test to use Deployment Apr 18, 2024
@adambkaplan

Copy link
Copy Markdown
Contributor

🤔 Now I'm wondering why it took us this long to catch the change...

@sanchezl

Copy link
Copy Markdown
Contributor

/retest


g.By("waiting for jenkins deployment")
err = exutil.WaitForDeploymentConfig(oc.KubeClient(), oc.AppsClient().AppsV1(), oc.Namespace(), "jenkins", 1, false, oc)
err = exutil.WaitForDeploymentReady(oc, "jenkins", oc.Namespace())

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

In WaitForDeploymentReady, the context isn't wired through to the PollUntilContextTimeout's ConditionFunc. No rush to fix it now if the test is working, but FYI.

@openshift-ci

openshift-ci Bot commented Apr 18, 2024

Copy link
Copy Markdown
Contributor

@apoorvajagtap: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-ovn-single-node-serial ba2a8f0 link false /test e2e-aws-ovn-single-node-serial
ci/prow/e2e-gcp-ovn-builds ba2a8f0 link true /test e2e-gcp-ovn-builds
ci/prow/e2e-aws-ovn-single-node-upgrade ba2a8f0 link false /test e2e-aws-ovn-single-node-upgrade

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-trt-bot

Copy link
Copy Markdown

Job Failure Risk Analysis for sha: ba2a8f0

Job Name Failure Risk
pull-ci-openshift-origin-master-e2e-aws-ovn-single-node-serial Low
[sig-arch] events should not repeat pathologically for ns/openshift-etcd-operator
This test has passed 49.30% of 71 runs on jobs ['periodic-ci-openshift-release-master-nightly-4.16-e2e-aws-ovn-single-node-serial'] in the last 14 days.

@soltysh soltysh left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/hold
Where that jenkins-ephemeral is coming from, I checked both https://github.com/openshift/origin/blob/master/examples/jenkins/jenkins-ephemeral-template.json and https://github.com/openshift/origin/blob/master/test/extended/testdata/cmd/test/cmd/testdata/jenkins/jenkins-ephemeral-template.json both of these are creating DCs not deployment?

Alternatively, if you're changing to a deployment (if the above what I said is wrong), you'll need to update logs command in line 65, it currently requests logs -f dc/jenkins which if we're working with a deployment should be logs -f deploy/jenkins

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 19, 2024
@soltysh

soltysh commented Apr 19, 2024

Copy link
Copy Markdown
Contributor

Found the culprit in openshift/cluster-samples-operator#540, so yes, aside from that wait, you need to update that run as I mentioned above.

Also, given we have those two extra places which don't seem to be in-sync with what we have in samples, I'd suggest to sync them as well. To make them all consistent.

@soltysh

soltysh commented Apr 19, 2024

Copy link
Copy Markdown
Contributor

/approve cancel
Please ping me directly on slack once all of the above is fixed, I don't want to have a half baked solution.

@soltysh soltysh removed the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 19, 2024
@soltysh

soltysh commented Apr 19, 2024

Copy link
Copy Markdown
Contributor

Also DumpApplicationPodLogs works for DCs only, so that needs fixing as well.

@soltysh

soltysh commented Apr 19, 2024

Copy link
Copy Markdown
Contributor

Reverting changes to cluster-samples-operator in openshift/cluster-samples-operator#542

@adambkaplan

Copy link
Copy Markdown
Contributor

/close

Replacing with #28732

@openshift-ci openshift-ci Bot closed this Apr 22, 2024
@openshift-ci

openshift-ci Bot commented Apr 22, 2024

Copy link
Copy Markdown
Contributor

@adambkaplan: Closed this PR.

Details

In response to this:

/close

Replacing with #28732

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot

Copy link
Copy Markdown

@apoorvajagtap: This pull request references Jira Issue OCPBUGS-32293. The bug has been updated to no longer refer to the pull request using the external bug tracker.

Details

In response to this:

Failing Jenkins Pipeline test blocks multiple PRs from merging. As agreed, skipping these tests temporarily.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants