Skip to content

OCPBUGS-32293: Use In-Tree Template for Jenkins - #28732

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
adambkaplan:jenkinspipeline-in-tree-template
Apr 25, 2024
Merged

OCPBUGS-32293: Use In-Tree Template for Jenkins#28732
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
adambkaplan:jenkinspipeline-in-tree-template

Conversation

@adambkaplan

Copy link
Copy Markdown
Contributor

Update the JenkinsPipeline build strategy test to use an in-tree template to deploy Jenkins during the test setup phase. This decouples the test from the Samples Operator, which is officially deprecated in 4.16 and will be phased out in the future. Previously, this test implicitly relied on the Samples Operator to install the Jenkins ephemeral template.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Apr 22, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@adambkaplan: This pull request references Jira Issue OCPBUGS-32293, which is valid.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.16.0) matches configured target version for branch (4.16.0)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

Update the JenkinsPipeline build strategy test to use an in-tree template to deploy Jenkins during the test setup phase. This decouples the test from the Samples Operator, which is officially deprecated in 4.16 and will be phased out in the future. Previously, this test implicitly relied on the Samples Operator to install the Jenkins ephemeral template.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot openshift-ci-robot added the jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. label Apr 22, 2024
@adambkaplan

Copy link
Copy Markdown
Contributor Author

/assign @soltysh

/cc @fbm3307 @sayan-biswas

This will decouple the JenkinsPipeline build test test from the Samples Operator. I'm setting aside the fact that we technically called this thing deprecated in OCP 4.3!

@adambkaplan

Copy link
Copy Markdown
Contributor Author

/hold

This still has an implict dependency on the samples operator (Jenkins imagestream). Once I see the build test get back to "green," I'll fix that.

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 22, 2024
Update the JenkinsPipeline build strategy test to use an in-tree
template and imagestream to deploy Jenkins during the test setup phase.
This decouples the test from the Samples Operator, which is officially
deprecated in 4.16 and will be phased out in the future. Previously,
this test implicitly relied on the Samples Operator to install the
Jenkins ephemeral template with the Jenkins imagestream image.

Signed-off-by: Adam Kaplan <adam.kaplan@redhat.com>
@adambkaplan
adambkaplan force-pushed the jenkinspipeline-in-tree-template branch from 78c4509 to 2775d33 Compare April 22, 2024 21:12
@adambkaplan

Copy link
Copy Markdown
Contributor Author

/hold cancel

PR updated

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 22, 2024
@adambkaplan

Copy link
Copy Markdown
Contributor Author

/retest

Build suite tests are passing! 😃

@openshift-trt-bot

Copy link
Copy Markdown

Job Failure Risk Analysis for sha: 2775d33

Job Name Failure Risk
pull-ci-openshift-origin-master-e2e-gcp-ovn-rt-upgrade IncompleteTests
Tests for this run (26) are below the historical average (717): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-aws-ovn-single-node-serial Low
[sig-arch] events should not repeat pathologically for ns/openshift-etcd-operator
This test has passed 53.70% of 54 runs on jobs ['periodic-ci-openshift-release-master-nightly-4.16-e2e-aws-ovn-single-node-serial'] in the last 14 days.

@soltysh soltysh left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 25, 2024
@openshift-ci

openshift-ci Bot commented Apr 25, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: adambkaplan, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 25, 2024
@sayan-biswas

Copy link
Copy Markdown

/retest

@openshift-ci

openshift-ci Bot commented Apr 25, 2024

Copy link
Copy Markdown
Contributor

@adambkaplan: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-ovn-single-node-serial 2775d33 link false /test e2e-aws-ovn-single-node-serial
ci/prow/e2e-aws-ovn-single-node-upgrade 2775d33 link false /test e2e-aws-ovn-single-node-upgrade

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit d8b06ca into openshift:master Apr 25, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@adambkaplan: Jira Issue OCPBUGS-32293: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-32293 has been moved to the MODIFIED state.

Details

In response to this:

Update the JenkinsPipeline build strategy test to use an in-tree template to deploy Jenkins during the test setup phase. This decouples the test from the Samples Operator, which is officially deprecated in 4.16 and will be phased out in the future. Previously, this test implicitly relied on the Samples Operator to install the Jenkins ephemeral template.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@adambkaplan
adambkaplan deleted the jenkinspipeline-in-tree-template branch May 6, 2024 13:28
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants