Skip to content

NO-JIRA: bump openshift/kubernetes - #28712

Merged
openshift-merge-bot[bot] merged 4 commits into
openshift:masterfrom
sanchezl:bump-ok
Apr 23, 2024
Merged

NO-JIRA: bump openshift/kubernetes#28712
openshift-merge-bot[bot] merged 4 commits into
openshift:masterfrom
sanchezl:bump-ok

Conversation

@sanchezl

Copy link
Copy Markdown
Contributor
  • bump openshift/kubernetes
  • update vendor
  • react to bump
  • make update

@sanchezl sanchezl changed the title bump ok NO-JIRA: bump openshift/kubernetes Apr 16, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@sanchezl: This pull request explicitly references no jira issue.

Details

In response to this:

  • bump openshift/kubernetes
  • update vendor
  • react to bump
  • make update

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Apr 16, 2024
@openshift-ci
openshift-ci Bot requested review from deads2k and soltysh April 16, 2024 03:12
@openshift-ci openshift-ci Bot added e2e-images-update Related to images used by e2e tests vendor-update Touching vendor dir or related files labels Apr 16, 2024
@sanchezl

Copy link
Copy Markdown
Contributor Author

/retest

2 similar comments
@soltysh

soltysh commented Apr 16, 2024

Copy link
Copy Markdown
Contributor

/retest

@soltysh

soltysh commented Apr 16, 2024

Copy link
Copy Markdown
Contributor

/retest

@openshift-trt-bot

Copy link
Copy Markdown

Job Failure Risk Analysis for sha: d813edc

Job Name Failure Risk
pull-ci-openshift-origin-master-e2e-agnostic-ovn-cmd High
[sig-arch][Late][Jira:"kube-apiserver"] collect certificate data [Suite:openshift/conformance/parallel]
This test has passed 99.88% of 2600 runs on release 4.16 [Overall] in the last week.

Open Bugs
openshift-tests image unauthorized
Help needed understanding certificate data collection problem on azure.
pull-ci-openshift-origin-master-e2e-metal-ipi-sdn IncompleteTests
Tests for this run (23) are below the historical average (1324): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-aws-ovn-fips IncompleteTests
Tests for this run (98) are below the historical average (1431): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)

@sanchezl

Copy link
Copy Markdown
Contributor Author

/retest

@soltysh soltysh left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@soltysh

soltysh commented Apr 18, 2024

Copy link
Copy Markdown
Contributor

/retest-required

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 18, 2024
@openshift-ci

openshift-ci Bot commented Apr 18, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: sanchezl, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 18, 2024
@sanchezl

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 886b7ae and 2 for PR HEAD 9daac42 in total

@sanchezl

Copy link
Copy Markdown
Contributor Author

e2e-gcp-ovn-builds needs #28718

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 5ac6743 and 1 for PR HEAD 9daac42 in total

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 783eb61 and 0 for PR HEAD 9daac42 in total

@soltysh

soltysh commented Apr 19, 2024

Copy link
Copy Markdown
Contributor

/override ci/prow/e2e-gcp-ovn-builds
this is being fixed in #28718

@openshift-ci

openshift-ci Bot commented Apr 19, 2024

Copy link
Copy Markdown
Contributor

@soltysh: Overrode contexts on behalf of soltysh: ci/prow/e2e-gcp-ovn-builds

Details

In response to this:

/override ci/prow/e2e-gcp-ovn-builds
this is being fixed in #28718

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 20, 2024
@soltysh

soltysh commented Apr 20, 2024

Copy link
Copy Markdown
Contributor

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 20, 2024
@soltysh

soltysh commented Apr 20, 2024

Copy link
Copy Markdown
Contributor

/retest-required

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 28ea635 and 2 for PR HEAD 9daac42 in total

@sanchezl

Copy link
Copy Markdown
Contributor Author

/retest-required

1 similar comment
@soltysh

soltysh commented Apr 21, 2024

Copy link
Copy Markdown
Contributor

/retest-required

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD dc09d1e and 1 for PR HEAD 9daac42 in total

@openshift-trt-bot

Copy link
Copy Markdown

Job Failure Risk Analysis for sha: 9daac42

Job Name Failure Risk
pull-ci-openshift-origin-master-e2e-metal-ipi-ovn-ipv6 IncompleteTests
Tests for this run (23) are below the historical average (937): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)

1 similar comment
@openshift-trt-bot

Copy link
Copy Markdown

Job Failure Risk Analysis for sha: 9daac42

Job Name Failure Risk
pull-ci-openshift-origin-master-e2e-metal-ipi-ovn-ipv6 IncompleteTests
Tests for this run (23) are below the historical average (937): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)

@soltysh

soltysh commented Apr 22, 2024

Copy link
Copy Markdown
Contributor

/retest-required

@soltysh

soltysh commented Apr 22, 2024

Copy link
Copy Markdown
Contributor

/test e2e-metal-ipi-ovn-ipv6

@soltysh

soltysh commented Apr 22, 2024

Copy link
Copy Markdown
Contributor

We got green run in metal in https://prow.ci.openshift.org/view/gs/test-platform-results/pr-logs/pull/28712/pull-ci-openshift-origin-master-e2e-metal-ipi-ovn-ipv6/1781727551937318912 and the current failures clearly point to infra problems.

/override ci/prow/e2e-metal-ipi-ovn-ipv6

@openshift-ci

openshift-ci Bot commented Apr 22, 2024

Copy link
Copy Markdown
Contributor

@soltysh: Overrode contexts on behalf of soltysh: ci/prow/e2e-metal-ipi-ovn-ipv6

Details

In response to this:

We got green run in metal in https://prow.ci.openshift.org/view/gs/test-platform-results/pr-logs/pull/28712/pull-ci-openshift-origin-master-e2e-metal-ipi-ovn-ipv6/1781727551937318912 and the current failures clearly point to infra problems.

/override ci/prow/e2e-metal-ipi-ovn-ipv6

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-trt-bot

Copy link
Copy Markdown

Job Failure Risk Analysis for sha: 9daac42

Job Name Failure Risk
pull-ci-openshift-origin-master-e2e-metal-ipi-ovn-ipv6 IncompleteTests
Tests for this run (13) are below the historical average (918): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-aws-ovn-serial Medium
[bz-networking][invariant] alert/OVNKubernetesResourceRetryFailure should not be at or above info
This test has passed 96.55% of 29 runs on jobs ['periodic-ci-openshift-release-master-nightly-4.16-e2e-aws-ovn-serial' 'periodic-ci-openshift-release-master-ci-4.16-e2e-aws-ovn-serial'] in the last 14 days.

@openshift-trt-bot

Copy link
Copy Markdown

Job Failure Risk Analysis for sha: 9daac42

Job Name Failure Risk
pull-ci-openshift-origin-master-e2e-metal-ipi-ovn-ipv6 IncompleteTests
Tests for this run (13) are below the historical average (920): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)

@soltysh

soltysh commented Apr 22, 2024

Copy link
Copy Markdown
Contributor

/retest-required

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD e06689a and 0 for PR HEAD 9daac42 in total

@openshift-trt-bot

Copy link
Copy Markdown

Job Failure Risk Analysis for sha: 9daac42

Job Name Failure Risk
pull-ci-openshift-origin-master-e2e-metal-ipi-ovn-ipv6 IncompleteTests
Tests for this run (13) are below the historical average (915): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)
pull-ci-openshift-origin-master-e2e-aws-ovn-serial Medium
[bz-networking][invariant] alert/OVNKubernetesResourceRetryFailure should not be at or above info
This test has passed 96.55% of 29 runs on jobs ['periodic-ci-openshift-release-master-nightly-4.16-e2e-aws-ovn-serial' 'periodic-ci-openshift-release-master-ci-4.16-e2e-aws-ovn-serial'] in the last 14 days.

@openshift-trt-bot

Copy link
Copy Markdown

Job Failure Risk Analysis for sha: 9daac42

Job Name Failure Risk
pull-ci-openshift-origin-master-e2e-metal-ipi-ovn-ipv6 IncompleteTests
Tests for this run (13) are below the historical average (915): IncompleteTests (not enough tests ran to make a reasonable risk analysis; this could be due to infra, installation, or upgrade problems)

@openshift-ci-robot

Copy link
Copy Markdown

/hold

Revision 9daac42 was retested 3 times: holding

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 23, 2024
@sanchezl

Copy link
Copy Markdown
Contributor Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 23, 2024
@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD c3ef267 and 2 for PR HEAD 9daac42 in total

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 4f37255 and 1 for PR HEAD 9daac42 in total

@openshift-ci

openshift-ci Bot commented Apr 23, 2024

Copy link
Copy Markdown
Contributor

@sanchezl: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-ovn-single-node-upgrade 9daac42 link false /test e2e-aws-ovn-single-node-upgrade

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@soltysh

soltysh commented Apr 23, 2024

Copy link
Copy Markdown
Contributor

/retest-required

@openshift-merge-bot
openshift-merge-bot Bot merged commit b2195bf into openshift:master Apr 23, 2024
@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

This PR has been included in build openshift-enterprise-tests-container-v4.17.0-202404231810.p0.gb2195bf.assembly.stream.el9 for distgit openshift-enterprise-tests.
All builds following this will include this PR.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. e2e-images-update Related to images used by e2e tests jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. vendor-update Touching vendor dir or related files

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants