WRKLDS-1251: Remove temporarily must-gather-base image from oc presubmits - #51578
Conversation
|
@ardaguclu: This pull request references WRKLDS-1251 which is a valid jira issue. Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the task to target the "4.16.0" version, but no target version was set. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
[REHEARSALNOTIFIER]
A total of 43 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs. A full list of affected jobs can be found here Interacting with pj-rehearseComment: Once you are satisfied with the results of the rehearsals, comment: |
|
/pj-rehearse pull-ci-openshift-oc-release-4.16-images |
|
@ardaguclu: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel. |
|
@ardaguclu: requesting more than one rehearsal in one comment is not supported. If you would like to rehearse multiple specific jobs, please separate the job names by a space in a single command. |
1 similar comment
|
@ardaguclu: requesting more than one rehearsal in one comment is not supported. If you would like to rehearse multiple specific jobs, please separate the job names by a space in a single command. |
|
/pj-rehearse pull-ci-openshift-oc-release-4.16-e2e-agnostic-ovn-cmd pull-ci-openshift-oc-release-4.16-e2e-aws-ovn |
|
@ardaguclu: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel. |
|
Worth trying |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: ardaguclu, ingvagabund The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/hold cancel |
|
I've verified that the must-gather imagestreamtag in CI system hasn't been overridden by cli image. |
|
Failure of https://prow.ci.openshift.org/view/gs/test-platform-results/pr-logs/pull/openshift_release/51578/rehearse-51578-pull-ci-openshift-oc-release-4.16-e2e-aws-ovn/1785942839276867584 is totally due to my creation of test pods to troubleshoot by directly accessing to the cluster and forgot to delete it. |
|
/pj-rehearse ack |
|
@ardaguclu: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel. |
|
/retest-required |
|
@ardaguclu: The following test failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
must-gather base image is overridden by the latest version of cli image in oc presubmits jobs due to this configuration. However, since we are trying to change the base image of oc to RHEL9 openshift/oc#1752 and must-gather is still on RHEL8 (oc.rhel9 does not work on RHEL8 hosts), we have to disable this override temporarily. Because we are explicitly stating to must-gather that is to use oc.rhel8 in openshift/must-gather#418 and openshift/must-gather#420.
this PR has to be reverted back;