Fix cli, cli-artifacts image generation for okd scos - #51533
Conversation
|
/pj-rehearse pull-ci-openshift-oc-master-okd-scos-images |
|
@Prashanth684: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel. |
|
/test ci-operator-registry |
|
/cc @LorbusChris |
|
/lgtm |
|
/pj-rehearse ack |
|
@LorbusChris: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel. |
| namespace: ocp | ||
| tag: rhel-9-golang-1.20-openshift-4.16 | ||
| ocp_builder_rhel-9-golang-1.21-openshift-4.16: | ||
| name: builder |
There was a problem hiding this comment.
@jupierce is this something that needs to be corrected in the automation, i.e the generation of rhel9 golang-1.21 builder images
|
/cc @soltysh |
| from: cli | ||
| inputs: | ||
| ocp_builder_rhel-8-golang-1.20-openshift-4.16: | ||
| ocp_builder_rhel-8-golang-1.21-openshift-4.16: |
There was a problem hiding this comment.
Bumping to go 1.21 makes sense. However, we have switched to RHEL9 entirely openshift/oc#1752 and I think, this configuration files should be simplified by only having rhel9-1-21 builder and removing the rest. Something similar to #51610
|
@Prashanth684, Interacting with pj-rehearseComment: Once you are satisfied with the results of the rehearsals, comment: |
In an effort to build OKD on top of centos stream(https://issues.redhat.com/browse/OKD-210), OKD-scos configs were added in openshift#50118. We are going through all images and identifying the ones which do not build and are working to fix them. Changed the golang-builder to use 1.21.
|
[REHEARSALNOTIFIER]
Interacting with pj-rehearseComment: Once you are satisfied with the results of the rehearsals, comment: |
|
/pj-rehearse pull-ci-openshift-oc-master-okd-scos-images |
|
@Prashanth684: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel. |
|
/lgtm |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: ardaguclu, LorbusChris, Prashanth684 The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/pj-rehearse ack |
|
@Prashanth684: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel. |
|
@Prashanth684: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
In an effort to build OKD on top of centos stream(https://issues.redhat.com/browse/OKD-210), OKD-scos configs were added in openshift#50118. We are going through all images and identifying the ones which do not build and are working to fix them. Changed the golang-builder to use 1.21.
In an effort to build OKD on top of centos stream(https://issues.redhat.com/browse/OKD-210), OKD-scos configs were added in openshift#50118. We are going through all images and identifying the ones which do not build and are working to fix them. Changed the golang-builder to use 1.21.
In an effort to build OKD on top of centos stream(https://issues.redhat.com/browse/OKD-210), OKD-scos configs were added in #50118. We are going through all images and identifying the ones which do not build and are working to fix them. Changed the golang-builder to use 1.21.