Skip to content

WRKLDS-1251: Use oc rhel8 binary explicitly until base image is rhel9 - #418

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
ardaguclu:use-oc-rhel8-explicitly
Apr 30, 2024
Merged

WRKLDS-1251: Use oc rhel8 binary explicitly until base image is rhel9#418
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
ardaguclu:use-oc-rhel8-explicitly

Conversation

@ardaguclu

Copy link
Copy Markdown
Member

must-gather includes oc invocations to gather some data and must-gather gets this oc from cli image. However, when we try to bump cli image to RHEL9, since must-gather is still on RHEL8, all oc invocations failed. This PR changes to use cli-artifacts image instead of cli. Thanks to that change, must-gather can explicitly uses oc.rhel8 and that decouples the cli and must-gather images for the future similar migrations.

example failures;

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Apr 29, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@ardaguclu: This pull request explicitly references no jira issue.

Details

In response to this:

must-gather includes oc invocations to gather some data and must-gather gets this oc from cli image. However, when we try to bump cli image to RHEL9, since must-gather is still on RHEL8, all oc invocations failed. This PR changes to use cli-artifacts image instead of cli. Thanks to that change, must-gather can explicitly uses oc.rhel8 and that decouples the cli and must-gather images for the future similar migrations.

example failures;

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested review from sferich888 and soltysh April 29, 2024 09:46
@ardaguclu

Copy link
Copy Markdown
Member Author

This PR depends on openshift/release#51449

@ardaguclu
ardaguclu force-pushed the use-oc-rhel8-explicitly branch from aa2ec10 to dd507c9 Compare April 29, 2024 10:59
@sferich888

Copy link
Copy Markdown
Contributor

While I don't like doing this; this is temporary (to work around SSL compatibility issue with FIPS across OCP versions).

/lgtm

@ardaguclu

Copy link
Copy Markdown
Member Author

/hold
until dependant merges

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 29, 2024
@openshift-ci openshift-ci Bot added lgtm Indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Apr 29, 2024
@openshift-ci

openshift-ci Bot commented Apr 29, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

joepvd added a commit to joepvd/ocp-build-data that referenced this pull request Apr 29, 2024
@joepvd

joepvd commented Apr 29, 2024

Copy link
Copy Markdown

/lgtm

@openshift-ci

openshift-ci Bot commented Apr 29, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ardaguclu, joepvd, sferich888

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ardaguclu

Copy link
Copy Markdown
Member Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 30, 2024
@openshift-merge-bot
openshift-merge-bot Bot merged commit 68204b5 into openshift:master Apr 30, 2024
@ardaguclu
ardaguclu deleted the use-oc-rhel8-explicitly branch April 30, 2024 10:40
joepvd added a commit to joepvd/ocp-build-data that referenced this pull request Apr 30, 2024
openshift-cherrypick-robot pushed a commit to openshift-cherrypick-robot/ocp-build-data that referenced this pull request Apr 30, 2024
@ardaguclu ardaguclu changed the title NO-JIRA: Use oc rhel8 binary explicitly until base image is rhel9 WRKLDS-1251: Use oc rhel8 binary explicitly until base image is rhel9 Apr 30, 2024
@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

This PR has been included in build ose-must-gather-container-v4.17.0-202404301246.p0.g68204b5.assembly.stream.el8 for distgit ose-must-gather.
All builds following this will include this PR.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants