Skip to content

must-gather inherits from cli-artifacts - #4754

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift-eng:openshift-4.16from
joepvd:must-gather-from-artifacts
Apr 30, 2024
Merged

must-gather inherits from cli-artifacts#4754
openshift-merge-bot[bot] merged 1 commit into
openshift-eng:openshift-4.16from
joepvd:must-gather-from-artifacts

Conversation

@joepvd

@joepvd joepvd commented Apr 29, 2024

Copy link
Copy Markdown
Contributor

@joepvd

joepvd commented Apr 29, 2024

Copy link
Copy Markdown
Contributor Author

/cherrypick openshift-4.17

@openshift-cherrypick-robot

Copy link
Copy Markdown

@joepvd: once the present PR merges, I will cherry-pick it on top of openshift-4.17 in a new PR and assign it to you.

Details

In response to this:

/cherrypick openshift-4.17

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Comment thread images/ose-must-gather.yml Outdated
builder:
- stream: golang
member: openshift-enterprise-cli
member: ose-cli-artifacts

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Do we still need this change considering that must-gather still bases on cli image and only extract oc.rhel8 from cli-artifacts in openshift/must-gather#418

@joepvd
joepvd force-pushed the must-gather-from-artifacts branch from 27f8fc3 to f8e482d Compare April 30, 2024 10:57
@ardaguclu

Copy link
Copy Markdown

/lgtm

@openshift-ci

openshift-ci Bot commented Apr 30, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: changing LGTM is restricted to collaborators

Details

In response to this:

/lgtm

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@joepvd joepvd added lgtm Indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Apr 30, 2024
@openshift-ci

openshift-ci Bot commented Apr 30, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by:

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit 1fde299 into openshift-eng:openshift-4.16 Apr 30, 2024
@openshift-cherrypick-robot

Copy link
Copy Markdown

@joepvd: new pull request created: #4760

Details

In response to this:

/cherrypick openshift-4.17

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants