Skip to content

WRKLDS-1251: Use oc.rhel8 based on the target arch of image - #420

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
ardaguclu:target-base-oc
May 2, 2024
Merged

WRKLDS-1251: Use oc.rhel8 based on the target arch of image#420
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
ardaguclu:target-base-oc

Conversation

@ardaguclu

@ardaguclu ardaguclu commented May 2, 2024

Copy link
Copy Markdown
Member

This PR uses the correct oc.rhel8 binary based on the image's architecture. This is the follow up of #418

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label May 2, 2024
@openshift-ci-robot

openshift-ci-robot commented May 2, 2024

Copy link
Copy Markdown

@ardaguclu: This pull request references WRKLDS-1251 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the task to target the "4.16.0" version, but no target version was set.

Details

In response to this:

This PR uses the correct oc.rhel8 binary based on the image's architecture.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested review from sferich888 and soltysh May 2, 2024 06:42
@joepvd

joepvd commented May 2, 2024

Copy link
Copy Markdown

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 2, 2024
@openshift-ci-robot

openshift-ci-robot commented May 2, 2024

Copy link
Copy Markdown

@ardaguclu: This pull request references WRKLDS-1251 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the task to target the "4.16.0" version, but no target version was set.

Details

In response to this:

This PR uses the correct oc.rhel8 binary based on the image's architecture. This is the follow up of #418

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci

openshift-ci Bot commented May 2, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@jeffdyoung

Copy link
Copy Markdown

/lgtm

@sferich888

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci

openshift-ci Bot commented May 2, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ardaguclu, jeffdyoung, joepvd, sferich888

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 2, 2024
@openshift-merge-bot
openshift-merge-bot Bot merged commit ff904a4 into openshift:master May 2, 2024
@ardaguclu
ardaguclu deleted the target-base-oc branch May 2, 2024 11:01
@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

This PR has been included in build ose-must-gather-container-v4.16.0-202405021515.p0.gff904a4.assembly.stream.el8 for distgit ose-must-gather.
All builds following this will include this PR.

@jaypoulz

jaypoulz commented May 2, 2024

Copy link
Copy Markdown

Should this have included s390x as well?

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants