Skip to content

NO-JIRA: copy oc binary from cli-artifacts - #419

Closed
jeffdyoung wants to merge 1 commit into
openshift:masterfrom
jeffdyoung:multi-arch
Closed

NO-JIRA: copy oc binary from cli-artifacts#419
jeffdyoung wants to merge 1 commit into
openshift:masterfrom
jeffdyoung:multi-arch

Conversation

@jeffdyoung

Copy link
Copy Markdown

additional work for #418 to copy multi-arch oc binaries.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label May 1, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@jeffdyoung: This pull request explicitly references no jira issue.

Details

In response to this:

additional work for #418 to copy multi-arch oc binaries.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested review from sferich888 and soltysh May 1, 2024 20:18
Comment thread Dockerfile.rhel7 Outdated
@sferich888

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 1, 2024
@openshift-ci

openshift-ci Bot commented May 1, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jeffdyoung, sferich888

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 1, 2024
@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 68204b5 and 2 for PR HEAD 9dda729 in total

@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label May 2, 2024
@openshift-ci

openshift-ci Bot commented May 2, 2024

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

Comment thread Dockerfile.rhel7
FROM registry.ci.openshift.org/ocp/4.16:cli
COPY --from=builder /go/src/github.com/openshift/must-gather/collection-scripts/* /usr/bin/
COPY --from=cli-artifacts /usr/share/openshift/linux_amd64/oc.rhel8 /usr/bin/oc
COPY --from=cli-artifacts /usr/bin/oc /usr/bin/oc

Copy link
Copy Markdown
Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@sferich888 we don't have to copy a specific oc arch here. The cli-artifacts container has the rhel8 variant of oc in /usr/bin/oc. Since the build system builds on native HW, we will always get the right binary arch.

https://github.com/openshift/oc/blob/master/images/cli-artifacts/Dockerfile.rhel#L15-L21

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

We are trying to change the default oc's base image to RHEL9 and default oc in cli-artifacts will also not work on must-gather.
/hold

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Proposed this #420 instead

@jeffdyoung jeffdyoung changed the title NO-Jira: copy multi-arch binary NO-JIRA: copy oc binary from cli-artifacts May 2, 2024
@openshift-ci

openshift-ci Bot commented May 2, 2024

Copy link
Copy Markdown
Contributor

@jeffdyoung: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label May 2, 2024
@jeffdyoung jeffdyoung closed this May 2, 2024
@jeffdyoung

Copy link
Copy Markdown
Author

closing in favor of #420

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants