Skip to content

WIP: Test e2e jobs - #4007

Closed
ardaguclu wants to merge 1 commit into
openshift:masterfrom
ardaguclu:test-e2e
Closed

WIP: Test e2e jobs#4007
ardaguclu wants to merge 1 commit into
openshift:masterfrom
ardaguclu:test-e2e

Conversation

@ardaguclu

Copy link
Copy Markdown
Member

Simple test to see that current jobs are working properly.

@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 4, 2024
@openshift-ci
openshift-ci Bot requested review from danilo-gemoli and smg247 March 4, 2024 10:24
@openshift-ci

openshift-ci Bot commented Mar 4, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: ardaguclu
Once this PR has been reviewed and has the lgtm label, please assign droslean for approval. For more information see the Kubernetes Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

/test remaining-required

@openshift-ci

openshift-ci Bot commented Mar 4, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e 2520b4e link true /test e2e
ci/prow/security 2520b4e link false /test security

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@ardaguclu

Copy link
Copy Markdown
Member Author

/close
verified that e2e is broken.

@openshift-ci openshift-ci Bot closed this Mar 4, 2024
@openshift-ci

openshift-ci Bot commented Mar 4, 2024

Copy link
Copy Markdown
Contributor

@ardaguclu: Closed this PR.

Details

In response to this:

/close
verified that e2e is broken.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@ardaguclu
ardaguclu deleted the test-e2e branch March 4, 2024 11:28
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants