Skip to content

OCPBUGS-20097: Switch openshift/kubernetes to golang 1.21 - #47661

Closed
soltysh wants to merge 1 commit into
openshift:masterfrom
soltysh:golang_1.21
Closed

OCPBUGS-20097: Switch openshift/kubernetes to golang 1.21#47661
soltysh wants to merge 1 commit into
openshift:masterfrom
soltysh:golang_1.21

Conversation

@soltysh

@soltysh soltysh commented Jan 15, 2024

Copy link
Copy Markdown
Contributor

/hold

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jan 15, 2024
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Jan 15, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@soltysh: This pull request references Jira Issue OCPBUGS-20097, which is valid.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.16.0) matches configured target version for branch (4.16.0)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

/hold

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot openshift-ci-robot added the jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. label Jan 15, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@soltysh: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-kubernetes-master-artifacts openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-configmap-scale openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-agnostic-ovn-cmd openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws-crun-wasm openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws-csi openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws-disruptive openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws-jenkins openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws-multitenant openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws-ovn openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws-ovn-cgroupsv2 openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws-ovn-crun openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws-ovn-downgrade openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws-ovn-fips openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws-ovn-serial openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws-ovn-upgrade openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-aws-single-node openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-azure openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-azure-ovn-upgrade openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-gcp openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-gcp-ovn-upgrade openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-metal-ipi openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-metal-ipi-ovn-dualstack openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-metal-ipi-ovn-ipv6 openshift/kubernetes presubmit Ci-operator config changed
pull-ci-openshift-kubernetes-master-e2e-openstack openshift/kubernetes presubmit Ci-operator config changed

A total of 40 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here
Prior to this PR being merged, you will need to either run and acknowledge or opt to skip these rehearsals.

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse abort to abort all active rehearsals

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@openshift-ci
openshift-ci Bot requested review from deads2k and mfojtik January 15, 2024 16:07
@openshift-ci

openshift-ci Bot commented Jan 15, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jan 15, 2024
@openshift-merge-robot openshift-merge-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Jan 28, 2024
@openshift-merge-robot

Copy link
Copy Markdown
Contributor

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-bot

Copy link
Copy Markdown
Contributor

Issues in openshift/release go stale after 30d of inactivity.

Mark the issue as fresh by commenting /remove-lifecycle stale.
Stale issues rot after an additional 15d of inactivity and eventually close.
Exclude this issue from closing by commenting /lifecycle frozen.

If this issue is safe to close now please do so with /close.

/lifecycle stale

@openshift-ci openshift-ci Bot added the lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. label Feb 27, 2024
@openshift-ci

openshift-ci Bot commented Mar 1, 2024

Copy link
Copy Markdown
Contributor

@soltysh: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/ci-operator-config-metadata a6cb5ba link true /test ci-operator-config-metadata
ci/prow/ci-operator-config a6cb5ba link true /test ci-operator-config

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-bot

Copy link
Copy Markdown
Contributor

Stale issue in openshift/release rot after 15d of inactivity.

Mark the issue as fresh by commenting /remove-lifecycle rotten.
Rotten issues close after an additional 15d of inactivity.
Exclude this issue from closing by commenting /lifecycle frozen.

If this issue is safe to close now please do so with /close.

/lifecycle rotten
/remove-lifecycle stale

@openshift-ci openshift-ci Bot added lifecycle/rotten Denotes an issue or PR that has aged beyond stale and will be auto-closed. and removed lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. labels Mar 17, 2024
@openshift-bot

Copy link
Copy Markdown
Contributor

Rotten issues in openshift/release close after 15d of inactivity.

Reopen the issue by commenting /reopen.
Mark the issue as fresh by commenting /remove-lifecycle rotten.
Exclude this issue from closing again by commenting /lifecycle frozen.

/close

@openshift-ci openshift-ci Bot closed this Apr 1, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@soltysh: This pull request references Jira Issue OCPBUGS-20097. The bug has been updated to no longer refer to the pull request using the external bug tracker.

Details

In response to this:

/hold

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci

openshift-ci Bot commented Apr 1, 2024

Copy link
Copy Markdown
Contributor

@openshift-bot: Closed this PR.

Details

In response to this:

Rotten issues in openshift/release close after 15d of inactivity.

Reopen the issue by commenting /reopen.
Mark the issue as fresh by commenting /remove-lifecycle rotten.
Exclude this issue from closing again by commenting /lifecycle frozen.

/close

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@soltysh
soltysh deleted the golang_1.21 branch April 24, 2024 13:08
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lifecycle/rotten Denotes an issue or PR that has aged beyond stale and will be auto-closed. needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants