fix: SDK login methods not saving token - #31278
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: ebdef7e The changes in this PR will be included in the next version bump. This PR includes changesets to release 30 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov ReportAttention:
Additional details and impacted files@@ Coverage Diff @@
## release-6.5.1 #31278 +/- ##
=================================================
- Coverage 52.20% 52.15% -0.06%
=================================================
Files 812 812
Lines 15186 15190 +4
Branches 2857 2857
=================================================
- Hits 7928 7922 -6
- Misses 6821 6829 +8
- Partials 437 439 +2
Flags with carried forward coverage won't be shown. Click here to find out more. |
ggazzo
force-pushed
the
fix/sdkAccount
branch
from
December 19, 2023 19:58
ee5ab34 to
1dfc0f7
Compare
ggazzo
force-pushed
the
fix/sdkAccount
branch
from
December 19, 2023 20:00
1dfc0f7 to
1500cb5
Compare
ggazzo
marked this pull request as ready for review
December 19, 2023 20:01
gabriellsh
added a commit
that referenced
this pull request
Dec 21, 2023
…o develop * 'release-6.5.1' of github.com:RocketChat/Rocket.Chat: fix: Fix wrong callback usage on after/before user actions (#31264) fix: SDK login methods not saving token (#31278) fix: account verify email (#31227) fix: stop refetching banner data (#31235) chore: fix date operation
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
Issue(s)
Steps to test or reproduce
Further comments