Skip to content

fix: Fix wrong callback usage on after/before user actions - #31264

Merged
kodiakhq[bot] merged 3 commits into
release-6.5.1from
fix/bad-after-user-callbacks
Dec 20, 2023
Merged

fix: Fix wrong callback usage on after/before user actions#31264
kodiakhq[bot] merged 3 commits into
release-6.5.1from
fix/bad-after-user-callbacks

Conversation

@KevLehman

@KevLehman KevLehman commented Dec 18, 2023

Copy link
Copy Markdown
Member

SUP-436

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Dec 18, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ab16ded

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/ddp-client Patch
@rocket.chat/livechat Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/gazzodown Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/model-typings Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/models Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Dec 18, 2023

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Comparison is base (f6a525a) 52.11% compared to head (ab16ded) 52.10%.

Additional details and impacted files

Impacted file tree graph

@@                Coverage Diff                @@
##           release-6.5.1   #31264      +/-   ##
=================================================
- Coverage          52.11%   52.10%   -0.02%     
=================================================
  Files                812      811       -1     
  Lines              15190    15186       -4     
  Branches            2857     2857              
=================================================
- Hits                7916     7912       -4     
+ Misses              6835     6834       -1     
- Partials             439      440       +1     
Flag Coverage Δ
e2e 48.58% <ø> (-0.02%) ⬇️
unit 69.61% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

lmauromb
lmauromb previously approved these changes Dec 18, 2023
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Dec 20, 2023
@KevLehman
KevLehman marked this pull request as ready for review December 20, 2023 13:46
@KevLehman
KevLehman requested a review from a team as a code owner December 20, 2023 13:46
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Dec 20, 2023
@ggazzo

ggazzo commented Dec 20, 2023

Copy link
Copy Markdown
Member

please rebase to 6.5.1 branch

@KevLehman
KevLehman marked this pull request as draft December 20, 2023 14:02
@KevLehman
KevLehman force-pushed the fix/bad-after-user-callbacks branch from 07c1b40 to 4bed2b7 Compare December 20, 2023 14:11
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Dec 20, 2023
@KevLehman
KevLehman changed the base branch from develop to release-6.5.1 December 20, 2023 14:12
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Dec 20, 2023
@KevLehman
KevLehman marked this pull request as ready for review December 20, 2023 14:14
@KevLehman

Copy link
Copy Markdown
Member Author

@ggazzo , there you go

@casalsgh
casalsgh requested a review from lmauromb December 20, 2023 14:18
@kodiakhq
kodiakhq Bot merged commit b42aa49 into release-6.5.1 Dec 20, 2023
@kodiakhq
kodiakhq Bot deleted the fix/bad-after-user-callbacks branch December 20, 2023 14:58
gabriellsh added a commit that referenced this pull request Dec 21, 2023
…o develop

* 'release-6.5.1' of github.com:RocketChat/Rocket.Chat:
  fix: Fix wrong callback usage on after/before user actions (#31264)
  fix: SDK login methods not saving token (#31278)
  fix: account verify email (#31227)
  fix: stop refetching banner data (#31235)
  chore: fix date operation
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants