fix: stop refetching banner data - #31235
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 26ca8dc The changes in this PR will be included in the next version bump. This PR includes changesets to release 30 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
ggazzo
marked this pull request as ready for review
December 13, 2023 20:40
Codecov Report
Additional details and impacted files@@ Coverage Diff @@
## release-6.5.1 #31235 +/- ##
=================================================
+ Coverage 52.13% 52.16% +0.02%
=================================================
Files 808 808
Lines 15169 15169
Branches 2852 2852
=================================================
+ Hits 7909 7913 +4
+ Misses 6823 6819 -4
Partials 437 437
Flags with carried forward coverage won't be shown. Click here to find out more. |
tassoevan
previously approved these changes
Dec 14, 2023
gabriellsh
added a commit
that referenced
this pull request
Dec 21, 2023
…o develop * 'release-6.5.1' of github.com:RocketChat/Rocket.Chat: fix: Fix wrong callback usage on after/before user actions (#31264) fix: SDK login methods not saving token (#31278) fix: account verify email (#31227) fix: stop refetching banner data (#31235) chore: fix date operation
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
Issue(s)
Steps to test or reproduce
Further comments