Skip to content

Handling Azure Parameters in CI Env using templates. - #582

Merged
kaovilai merged 11 commits into
openshift:masterfrom
deepakraj1997:multi-cloud-template
Apr 29, 2022
Merged

kaovilai merged 11 commits into
openshift:masterfrom
deepakraj1997:multi-cloud-template

Conversation

@deepakraj1997

@deepakraj1997 deepakraj1997 commented Mar 2, 2022

Copy link
Copy Markdown
Contributor
  • Moved most of the parameters to templates and scripts
  • Removed redundant code and unused functions.

@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 2, 2022
@openshift-ci

openshift-ci Bot commented Mar 2, 2022

Copy link
Copy Markdown

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@deepakraj1997
deepakraj1997 changed the base branch from master to ci-multi-cloud March 2, 2022 18:54
@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/test all

@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/uncc alaypatel07 JaydipGabani

@deepakraj1997 deepakraj1997 changed the title Handling Azure Parameters in CI Env using templates. [WIP] Handling Azure Parameters in CI Env using templates. Mar 2, 2022
@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/test all

3 similar comments
@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/test all

@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/test all

@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/test all

@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/test operator-e2e-azure

5 similar comments
@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/test operator-e2e-azure

@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/test operator-e2e-azure

@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/test operator-e2e-azure

@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/test operator-e2e-azure

@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/test operator-e2e-azure

@deepakraj1997
deepakraj1997 changed the base branch from ci-multi-cloud to master March 29, 2022 17:03
@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/test all

@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/retest

@deepakraj1997
deepakraj1997 marked this pull request as ready for review March 31, 2022 16:01
@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/retest

@deepakraj1997 deepakraj1997 changed the title [WIP] Handling Azure Parameters in CI Env using templates. Handling Azure Parameters in CI Env using templates. Apr 5, 2022
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Apr 5, 2022
@weshayutin

Copy link
Copy Markdown
Contributor

/retest

1 similar comment
@kaovilai

Copy link
Copy Markdown
Member

/retest

@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/retest

2 similar comments
@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/retest

@deepakraj1997

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-ci

openshift-ci Bot commented Apr 21, 2022

Copy link
Copy Markdown

@deepakraj1997: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@weshayutin

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 22, 2022
@kaovilai

Copy link
Copy Markdown
Member

looks like I made a conflicting PR #651 heh

@kaovilai

Copy link
Copy Markdown
Member

If this can merge I'll merge this first.

@kaovilai
kaovilai merged commit 69ce063 into openshift:master Apr 29, 2022
kaovilai pushed a commit to kaovilai/oadp-operator that referenced this pull request May 9, 2022
* Adding azure parameters

* Adding Openshift CI params

* Update Test Suite and Helpers

* Adding cred ref

* Adding cred ref

* Changing azure oadp cred dir to tmp

* Fixing duplicate error

* Removing OpenShift CI

* Changing test instance name

* Changing the AWS BSL Profile to default

* Changing BslMountPath variable in registry controller
kaovilai pushed a commit to kaovilai/oadp-operator that referenced this pull request May 10, 2022
* Adding azure parameters

* Adding Openshift CI params

* Update Test Suite and Helpers

* Adding cred ref

* Adding cred ref

* Changing azure oadp cred dir to tmp

* Fixing duplicate error

* Removing OpenShift CI

* Changing test instance name

* Changing the AWS BSL Profile to default

* Changing BslMountPath variable in registry controller
dymurray pushed a commit that referenced this pull request May 10, 2022
…n updateRegistrySecret (#679)

* Replace carriage return when getting provider secrets for registry + E2E changes (#652)

* Replace carriage return when getting provider secrets for registry

Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com>

* add tests

* replaceCarriageReturn unit test

* Registry credentials can use other secret names

Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com>

* Multi cloud E2E Updates (#568)

* Adding changes from ci-multi-cloud branch

* Updating flags in makefile

* Updating flags in makefile

* Updating flags in makefile

* Update defaults

* Fix helpers

* Cleaning up makefile

* Update makefile

* Update kustomization

* Cleaning up Makefile

* Adding changes for AWS CredentialFile

* Handling no default backuplocations

* Removing comments

* Adding config check in helpers

* Fix AWS Test case

* Fix basic review comments

* genericTests...

* Changing Credentials Mount Name for BSL when CredentialsFile key is used in Config - Fixes CI (#637)

* Changing Credentials Mount Name for BSL for GCP

* Fix DefaultPlugin for credentialsFile

* E2E Tests: Fix azure templating name.  (#573)

* Fixing Azure to run locally

* Aligning json

* Handling Azure Parameters in CI Env using templates. (#582)

* Adding azure parameters

* Adding Openshift CI params

* Update Test Suite and Helpers

* Adding cred ref

* Adding cred ref

* Changing azure oadp cred dir to tmp

* Fixing duplicate error

* Removing OpenShift CI

* Changing test instance name

* Changing the AWS BSL Profile to default

* Changing BslMountPath variable in registry controller

* Replace carriage return when getting provider secrets for registry + E2E changes (#652)

* Replace carriage return when getting provider secrets for registry

Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com>

* add tests

* replaceCarriageReturn unit test

* Registry credentials can use other secret names

Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com>

* fix missing err return for updateRegistrySecret function

* go fmt

* upgraded ginkgo on tests to ginkgo 2.0 (#560)

Co-authored-by: Nitish Srivastava <43022982+nitishSr@users.noreply.github.com>
Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com>
Co-authored-by: Deepak Raj D S <drajds@redhat.com>
Co-authored-by: Shubham Pampattiwar <shubhampampattiwar7@gmail.com>
Co-authored-by: Maya Peretz <mperetz@redhat.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

lgtm Indicates that a PR is ready to be merged. testing

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants