Skip to content

[OADP-463] cherrypicks 1.0 - carriage return + #678 missing err return updateRegistrySecret - #679

Merged
dymurray merged 10 commits into
openshift:oadp-1.0from
kaovilai:OADP-463-cp
May 10, 2022
Merged

dymurray merged 10 commits into
openshift:oadp-1.0from
kaovilai:OADP-463-cp

Conversation

@kaovilai

@kaovilai kaovilai commented May 9, 2022

Copy link
Copy Markdown
Member

[ OADP-463 ] cherrypicks 1.0 + #678

Cherrypicking #652 for 1.0 requires
#637 which requires
#568 which requires
#560 which requires
#531

Cherrypick additionally #678

@openshift-ci
openshift-ci Bot requested review from jmontleon and sseago May 9, 2022 17:49
@kaovilai kaovilai changed the title [OADP-463] cherrypicks 1.0 + #678 [OADP-463] cherrypicks 1.0 - carriage return + #678 missing err return updateRegistrySecret May 9, 2022
@dymurray

dymurray commented May 9, 2022

Copy link
Copy Markdown
Member

/retest

nitishSr and others added 10 commits May 10, 2022 10:54
…E2E changes (openshift#652)

* Replace carriage return when getting provider secrets for registry

Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com>

* add tests

* replaceCarriageReturn unit test

* Registry credentials can use other secret names

Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com>
* Adding changes from ci-multi-cloud branch

* Updating flags in makefile

* Updating flags in makefile

* Updating flags in makefile

* Update defaults

* Fix helpers

* Cleaning up makefile

* Update makefile

* Update kustomization

* Cleaning up Makefile

* Adding changes for AWS CredentialFile

* Handling no default backuplocations

* Removing comments

* Adding config check in helpers

* Fix AWS Test case

* Fix basic review comments
…sed in Config - Fixes CI (openshift#637)

* Changing Credentials Mount Name for BSL for GCP

* Fix DefaultPlugin for credentialsFile
* Fixing Azure to run locally

* Aligning json
* Adding azure parameters

* Adding Openshift CI params

* Update Test Suite and Helpers

* Adding cred ref

* Adding cred ref

* Changing azure oadp cred dir to tmp

* Fixing duplicate error

* Removing OpenShift CI

* Changing test instance name

* Changing the AWS BSL Profile to default

* Changing BslMountPath variable in registry controller
…E2E changes (openshift#652)

* Replace carriage return when getting provider secrets for registry

Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com>

* add tests

* replaceCarriageReturn unit test

* Registry credentials can use other secret names

Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com>
@openshift-ci

openshift-ci Bot commented May 10, 2022

Copy link
Copy Markdown

@kaovilai: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@kaovilai

Copy link
Copy Markdown
Member Author

AWESOMMMEEEEEEEEEE

@dymurray
dymurray merged commit ce423c4 into openshift:oadp-1.0 May 10, 2022
@kaovilai
kaovilai removed request for jmontleon and sseago May 12, 2022 16:16
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants