E2E Tests: Fix azure templating name. - #573
Merged
deepakraj1997 merged 2 commits intoFeb 21, 2022
Merged
Conversation
Member
|
What about here (switch/case)? Is the bool not working? case "azure": |
Contributor
Author
|
/retest |
|
@deepakraj1997: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
Contributor
Author
|
@kaovilai The bool is working perfectly fine. Its just that we need be able to test VSL and BSL in both CI and non-CI env hence i added those. |
kaovilai
approved these changes
Feb 21, 2022
kaovilai
pushed a commit
to kaovilai/oadp-operator
that referenced
this pull request
May 9, 2022
* Fixing Azure to run locally * Aligning json
kaovilai
pushed a commit
to kaovilai/oadp-operator
that referenced
this pull request
May 10, 2022
* Fixing Azure to run locally * Aligning json
dymurray
pushed a commit
that referenced
this pull request
May 10, 2022
…n updateRegistrySecret (#679) * Replace carriage return when getting provider secrets for registry + E2E changes (#652) * Replace carriage return when getting provider secrets for registry Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com> * add tests * replaceCarriageReturn unit test * Registry credentials can use other secret names Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com> * Multi cloud E2E Updates (#568) * Adding changes from ci-multi-cloud branch * Updating flags in makefile * Updating flags in makefile * Updating flags in makefile * Update defaults * Fix helpers * Cleaning up makefile * Update makefile * Update kustomization * Cleaning up Makefile * Adding changes for AWS CredentialFile * Handling no default backuplocations * Removing comments * Adding config check in helpers * Fix AWS Test case * Fix basic review comments * genericTests... * Changing Credentials Mount Name for BSL when CredentialsFile key is used in Config - Fixes CI (#637) * Changing Credentials Mount Name for BSL for GCP * Fix DefaultPlugin for credentialsFile * E2E Tests: Fix azure templating name. (#573) * Fixing Azure to run locally * Aligning json * Handling Azure Parameters in CI Env using templates. (#582) * Adding azure parameters * Adding Openshift CI params * Update Test Suite and Helpers * Adding cred ref * Adding cred ref * Changing azure oadp cred dir to tmp * Fixing duplicate error * Removing OpenShift CI * Changing test instance name * Changing the AWS BSL Profile to default * Changing BslMountPath variable in registry controller * Replace carriage return when getting provider secrets for registry + E2E changes (#652) * Replace carriage return when getting provider secrets for registry Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com> * add tests * replaceCarriageReturn unit test * Registry credentials can use other secret names Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com> * fix missing err return for updateRegistrySecret function * go fmt * upgraded ginkgo on tests to ginkgo 2.0 (#560) Co-authored-by: Nitish Srivastava <43022982+nitishSr@users.noreply.github.com> Co-authored-by: GitHub Co-pilot <idkwhatemailgithubcopilotuses@github.com> Co-authored-by: Deepak Raj D S <drajds@redhat.com> Co-authored-by: Shubham Pampattiwar <shubhampampattiwar7@gmail.com> Co-authored-by: Maya Peretz <mperetz@redhat.com>
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
I am in process of removing this completely and making it platform agnostic as per review made here: #568 (comment)
This is just a temporary patch